)]}'
{"id":"openstack%2Ftrove~727977","triplet_id":"openstack%2Ftrove~master~Ic81e52b5cf6c64d752358ab29a1f9a55bf25c88c","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ic81e52b5cf6c64d752358ab29a1f9a55bf25c88c","subject":"Remove translation sections from setup.cfg","status":"MERGED","created":"2020-05-14 08:57:33.000000000","updated":"2020-05-27 23:44:33.000000000","submitted":"2020-05-27 23:43:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727977-1590622994032-c44eae83","meta_rev_id":"ba4631334fdb979f8b542ae6bf6b250bfab829c0","_number":727977,"virtual_id_number":727977,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 23:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-27 08:31:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-27 20:34:20.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 20:34:20.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-05-27 23:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bcb4d6b0807a43cfbee741e362154e066fb71cc","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-05-14 08:57:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dca0a3e9243f82e89a5a019baaff04a890d652d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:34:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af31c9c586314b9db718195a694e4ecc : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/166564aa07174b169f924a7beb7a32e8 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac2c707151d548da94b72abdb0dc67ee : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41600aaf7ed44b84ba058c5cee12ee9f : FAILURE in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b01b93a56e24d1597058f89973e54bd : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f093e70ea02411fa233074115f2fc29 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc37933287844c685f39a8614062e6e : SUCCESS in 6m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c129895bc72f4dd084069b729ae8517c : SUCCESS in 10m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a609c4d88f204a36a8ba769cc65faba9 : SUCCESS in 7m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/11445c246d0d4f6098ecf9d0c308282a : FAILURE in 1h 25m 51s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f6c45b655f124794a4bb0368aede51b3 : SUCCESS in 2h 10m 04s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8bad63b7e8204f11a57e3d73f5daea2a : SUCCESS in 2h 14m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c80223e76e214aa989a8464507df5516 : SUCCESS in 1h 33m 01s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/69125c8670e84e3a9b8954c0a18517b0 : SUCCESS in 2h 28m 53s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8f290d2678754b20bb2d1def195d5205 : FAILURE in 46m 37s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/1245eec8118d49a6abc4fb03e8db530c : FAILURE in 45m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/80571e87eabe4f08a8b6358511f2a5ed : FAILURE in 1h 02m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13d0970c77f14073a82543028b5af33c : FAILURE in 1h 34m 41s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/0d1556dccab5473da03784a3f914509b : SUCCESS in 2h 06m 42s (non-voting)\n\nWarning:\n  Comments left for invalid file trove/limits/views.py\n  Comments left for invalid file trove/instance/views.py\n  Comments left for invalid file trove/tests/api/limits.py","accounts_in_message":[],"_revision_number":1},{"id":"151bccb51b790ea7f10e0b8b00e585dbfdd01f71","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-14 20:11:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b4e55c1010be00c72caef6f7fe05f66b3cf25ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 22:45:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/290670c1048648a59fcbdc8ebc71136a : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/162f98c7ee0444de9e712291352f15ad : FAILURE in 7m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd1b09a74ac24c7eb39715e43fb48779 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a0ef47a3ab4f2fab665cb0c4cf8090 : FAILURE in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a80a5d4f1cb4f469fc6b3ce52d5fd2a : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b2161bd5c9c44cb8fe1289500891a2e : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58ac0423e1604d3c91aff2f1642f8c58 : SUCCESS in 9m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0555271de1d40a481719a4f4f6e61f5 : SUCCESS in 9m 46s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12714683b71c4adba6cedf2dac2821ad : SUCCESS in 6m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/505a7de95a1942adb93d6128bae30761 : FAILURE in 1h 28m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7b8d35155a4146c5b9d3b117fb7c5c67 : SUCCESS in 2h 01m 08s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/53f341d72beb417e92202ecc412f832d : SUCCESS in 2h 25m 58s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b039700a5fc54f64917386830c003c98 : SUCCESS in 1h 33m 50s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f39efffc080b4ed3a0b9c59d1c475be7 : SUCCESS in 2h 32m 05s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/93cf1e8ab0814aa28f0535427525af7d : FAILURE in 41m 06s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f4e9f1112b2146808fcb7097675d257b : FAILURE in 40m 05s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/fac799020f584c899ee08d6d3709302e : FAILURE in 1h 17m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/361970f791394a668583ff07d9b3a959 : FAILURE in 1h 32m 33s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/98ac1d8dc07246b5b583cd7128bc9f06 : SUCCESS in 2h 01m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file trove/instance/views.py\n  Comments left for invalid file trove/tests/api/limits.py\n  Comments left for invalid file trove/limits/views.py","accounts_in_message":[],"_revision_number":1},{"id":"75bd0cafc9150a2d666148eadadee37397a84ee1","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-05-26 08:25:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff98430f80757b932700d51a44c8beb5f234a2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 10:42:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b975b152824d4fbe9ee7d9ba236d01c0 : SUCCESS in 3m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1e4d0df9c41646c9b565e7cfdb82b811 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95263796511248d69dd05f700b24487a : FAILURE in 6m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af0d35b941ff49d2bbaa5a54ce7b2abf : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21d55087d7a4407daf19dde7b2362a59 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b78eec4837d740aaaeded0ffeca6dc20 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a74192025264793894c0b8d438922d4 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818659c68fdf4f178ef4532252c5a021 : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d15b4929198b406fb6def3a11fd51378 : SUCCESS in 7m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3a15a9c5eb04e48ad74195f7f6c15bd : SUCCESS in 7m 22s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1b8af9fafd6c44bfb418527818426254 : SUCCESS in 57m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/469ccbeb2e8e42ef99bff8d879d4dc5e : SUCCESS in 2h 06m 53s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/25801d67e1fc46708fe46c99f17f1d53 : FAILURE in 1h 15m 46s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/90f41b4e86e34d68a469df599c40d098 : FAILURE in 1h 24m 17s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4b50d353cbf141a09cb9934460ab6038 : SUCCESS in 2h 15m 46s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/586e53dd68af4dca8727181660effea2 : FAILURE in 46m 47s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/c5d6bdd309c54f19882a5af98e39a4b6 : FAILURE in 57m 12s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f48a5f1c2a594071b2b5d6e01c43f9a5 : FAILURE in 1h 17m 10s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16fa17912dbf4c4ca00578baa238f6c1 : FAILURE in 1h 43m 13s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/4c09e89aa0654bf29b73db6482bac8bf : SUCCESS in 1h 54m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ebe085592f50506b987e9ed27c42e1fac66ec0a1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 01:28:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0650374734033e3890a2618be4d21301383cd699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 03:39:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1956ef90cbf14b5e8de13a7cf51d691b : SUCCESS in 3m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a1b32c94c2547e49b14d2b8f03038eb : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17a98149c70b4a3384636d1a22e97916 : FAILURE in 6m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/833347d18d90405c9b0f6baf0d6f4e7e : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65acb18e80034765b47f44c1ff9c12f2 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d0da17c702f4971bfe3a71f3143a279 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93bbdc11b270478ea5ae8c1d12ce5f7b : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258531433cac4d2289851e632cf4ff00 : SUCCESS in 4m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba227c5046084daca811eabc6b72d89d : SUCCESS in 8m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d327a8bb354b4c308a336a941fd65f13 : SUCCESS in 5m 20s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/07a8af983dfb4b9aa068dd9a0f0652d6 : SUCCESS in 48m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7453fca9dc8e40b29a28ca1454dbe775 : SUCCESS in 2h 05m 52s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5cd5ea0389974dac9c05061bc4b54a3b : FAILURE in 31m 22s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/9c55649e409949a5a2294dfb7dc2e09f : SUCCESS in 1h 25m 52s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/b1d6762ca6a44885b69a1cb3e1716103 : SUCCESS in 2h 04m 28s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8b0b8175622b49c381bb018661d5f146 : FAILURE in 37m 56s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/69902a0946f84606be99c7b22797ac0f : FAILURE in 35m 34s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/9857328731aa4c97b21b5814e5d32816 : FAILURE in 1h 04m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e262a39bb89b4be4aed73bef6867f4ca : SUCCESS in 52m 24s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/cb89c24baed5438aba14f24d39cff882 : SUCCESS in 1h 59m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"94b4c20187d83dd31ee33c02f756c194efd8332b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 08:31:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"95e633a5cb77819f1aa434ac9ddd6df9b500e50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:42:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56920aa1855147b989c6dae84c0d10da : SUCCESS in 4m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b62fc7c7a7f948238fb27ae27faf2948 : SUCCESS in 3m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bee0f72f61444b0a7428eccc00553a1 : FAILURE in 6m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99f5bb44306e4b8db31d342a14ec3c1e : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a5bff37df194133926d74e0dadce4ec : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed35ccc23c048dca1101b91d06215a2 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89bebd07c3a94c5b845c96c89ed30f5c : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/901321b3275c4a87b62dd3d4226ebe1c : SUCCESS in 7m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/89f45b749b0d4180ba393bf51c06d08e : SUCCESS in 8m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2f5597cee8c46e09052bd4f7c5a9ec5 : SUCCESS in 7m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/76de9023cabb49a689474072c6cb3665 : SUCCESS in 1h 08m 57s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d9d7bb7764024d1698cd797973be62af : TIMED_OUT in 3h 03m 39s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/52a084f21b4f4386910c4d47c0a3c667 : TIMED_OUT in 3h 05m 38s (non-voting)\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3dabd411006746aea4517e07c78d279b : FAILURE in 2h 15m 42s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e3225507fadd47fe9e1909f239dbf969 : FAILURE in 42m 40s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/7f474673c8504896a17c9b7be6d1fee8 : FAILURE in 40m 10s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f876764f3ec47358f8b855148b0e270 : FAILURE in 1h 18m 38s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/019eadfaf9ee45c2ae34748555c465a4 : FAILURE in 1h 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e83b7c2285bcc8f7984a5f8da36f006c85d63aa2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 20:34:20.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6c189b4720f470ee703011f2e0184d16550d86e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:34:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6872a68ab971807ee09444f0624fa07995140b5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:43:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eabf528619874f8fa1e8b872f092cd80 : SUCCESS in 4m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad352817dc564520bf6d4c9ba7168a57 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/129a0840bfe042888408635fbae26ddf : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e9ecf31dfe479c96641e5afebf7122 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91ce66e3b72949ed9e26f4599e909a58 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42cedc7313eb4449a5eda8371640c182 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df0f303cd2e4371826bd6db81fe4d77 : SUCCESS in 7m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/abea7d71187c45ffbe90646b85200e51 : SUCCESS in 9m 58s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3316e1247d2c4d74ad9fe72b67cc0c1e : TIMED_OUT in 3h 02m 46s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/00b6ac1b8f3646db9e54b5cd04793d52 : FAILURE in 1h 00m 26s (non-voting)\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/b47974ff49294c4d8d7f4878522c1c0b : FAILURE in 2h 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e23fbf29826501027fb8302a4cbadd8b1b644a69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:43:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ba4631334fdb979f8b542ae6bf6b250bfab829c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:44:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1555dcdd54cf4f3aa7a384b4d20330f4 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8cd3670bddd04cb9f71bbebf06d70fc8f55e3df4","revisions":{"54c8259927d5c2dfbd7a6aa2721ca7b50f1299ab":{"kind":"REWORK","_number":1,"created":"2020-05-14 08:57:33.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/77/727977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/727977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/727977/1"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 08:54:56.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 08:54:56.000000000","tz":480},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore, Babel can\ngenerate translation files without them.\n\nChange-Id: Ic81e52b5cf6c64d752358ab29a1f9a55bf25c88c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54c8259927d5c2dfbd7a6aa2721ca7b50f1299ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/54c8259927d5c2dfbd7a6aa2721ca7b50f1299ab"}]},"branch":"refs/heads/master"},"7cbbf83deae03f1fb381830b66287b30ca477e81":{"kind":"REWORK","_number":2,"created":"2020-05-26 08:25:15.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/77/727977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/727977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/727977/2"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 08:54:56.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-26 08:24:41.000000000","tz":480},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore, Babel can\ngenerate translation files without them.\n\nChange-Id: Ic81e52b5cf6c64d752358ab29a1f9a55bf25c88c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7cbbf83deae03f1fb381830b66287b30ca477e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7cbbf83deae03f1fb381830b66287b30ca477e81"}]},"branch":"refs/heads/master"},"8cd3670bddd04cb9f71bbebf06d70fc8f55e3df4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-27 08:31:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/77/727977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/727977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/727977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/727977/3"}}},"commit":{"parents":[{"commit":"91a6e5fa8aebfd579271a2f78536d526142d6228","subject":"Merge \"Update docs upgrade datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a6e5fa8aebfd579271a2f78536d526142d6228"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 08:54:56.000000000","tz":480},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 08:31:09.000000000","tz":0},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore, Babel can\ngenerate translation files without them.\n\nChange-Id: Ic81e52b5cf6c64d752358ab29a1f9a55bf25c88c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8cd3670bddd04cb9f71bbebf06d70fc8f55e3df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8cd3670bddd04cb9f71bbebf06d70fc8f55e3df4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
