)]}'
{"id":"openstack%2Ftrove~728816","triplet_id":"openstack%2Ftrove~stable%2Fussuri~Ia3999c0cfbd104f723eabdf08c842e3f8d671b7b","project":"openstack/trove","branch":"stable/ussuri","topic":"ussuri-branch-name","hashtags":[],"change_id":"Ia3999c0cfbd104f723eabdf08c842e3f8d671b7b","subject":"Use stable/ussuri for CI job and image build","status":"ABANDONED","created":"2020-05-18 12:00:28.000000000","updated":"2020-05-26 03:45:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ecfdd2c752f98c47b2e027d22e63b9254e31194","_number":728816,"virtual_id_number":728816,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-19 03:09:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 03:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07253dab7a6bbfff11592040a6b8d63e4f0e7321","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-18 12:00:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"987478bae2fce51d6211d1222bfe53fe70362d2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 13:32:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99e170e23dc74fca962c07dd350d919e : FAILURE in 7m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46636d07cc1d48c7aff8f358b5bdbef2 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52225dfccd874b3aaab717fae1a03d2a : FAILURE in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ecb88198eda46569492b5d80c079af1 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/055423af4d704905a6f88ce1cefde0cb : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b1ee96d1be4fa29b87c2a054f9f7c2 : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0fa3871b48e4b94950c4c0467838bc0 : SUCCESS in 8m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e3ec1f6c7f2c4f99b0d2f627105cfbdd : SUCCESS in 6m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0954b9f2b5f245ef86905e083d58e174 : SUCCESS in 1h 06m 49s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2720512232f1438d9f6c14862a238d72 : FAILURE in 44m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/56ac9696e0fa492094ee53856c2ff88f : FAILURE in 44m 08s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e61dc03c21ab4bcc8c4c7f56c9eb16e8 : FAILURE in 1h 01m 19s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4cf422e519f84cf29a21b1fa049e2420 : FAILURE in 49m 35s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/581403506446417c8866c5c3421af5a5 : FAILURE in 53m 47s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/35fddd85bb0d41b4819e6b3a853e19c1 : FAILURE in 42m 31s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f2241e7b10164e949a5bc58313b0d0f1 : FAILURE in 56m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b40e24b8c4c948109273376334275d3e : FAILURE in 1h 30m 47s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/1b552ca266574f62a8c8e2fac8b35d8b : FAILURE in 45m 18s (non-voting)\n\nWarning:\n  Comments left for invalid file trove/common/stream_codecs.py\n  Comments left for invalid file trove/common/apischema.py\n  Comments left for invalid file trove/common/context.py\n  Comments left for invalid file trove/tests/unittests/common/test_context.py\n  Comments left for invalid file trove/tests/unittests/extensions/common/test_service.py\n  Comments left for invalid file trove/db/sqlalchemy/session.py\n  Comments left for invalid file trove/common/notification.py\n  Comments left for invalid file trove/tests/unittests/common/test_wsgi.py\n  Comments left for invalid file trove/instance/views.py\n  Comments left for invalid file trove/tests/api/limits.py\n  Comments left for invalid file trove/limits/views.py","accounts_in_message":[],"_revision_number":1},{"id":"0b07d98be7f008df6f7259e99f6bd63e33f40fbc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-19 00:50:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d082edfd3c82150f91b3d3dec9ab0cdfb39a00b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-19 00:51:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"501c05ffd21d7ecf3e47c639c2c5b5c87ad0c1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 03:09:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fe4f723879247f299d6442f2be8e116 : FAILURE in 14m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e75fcfb7bf73437094e9d38442ee6fb9 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b3624672414c578c5421065285065f : FAILURE in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/370f615ddfdf498fbae682992a123d06 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7fd90ed08204b129f3d4d45145fb334 : SUCCESS in 12m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c51dc406454a9b8b4614fc25246c48 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4928384e762940b6a14e298f6e33a036 : SUCCESS in 12m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc2f528f759d466eb8037bdc88d3a9a7 : SUCCESS in 9m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/30547a100aad43749d8682887c1c3229 : SUCCESS in 52m 08s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/16051fe5f1954fdcaef62095071456ca : SUCCESS in 2h 08m 06s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/338aede4f17d4042b5079e9a5993bacd : SUCCESS in 2h 11m 16s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/282947dfe4ce471094b6c402606f3bab : SUCCESS in 1h 40m 46s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/cf0e591536924536bbee61a22211b8c7 : SUCCESS in 2h 15m 35s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/09d6e28b18304d70aacb5c1cf4310e82 : FAILURE in 40m 16s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/cb36cbaf591041c885cf7387dbe79104 : FAILURE in 50m 50s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/ce8c840874964e2a983e1a48bd65c9f3 : FAILURE in 1h 10m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6afd0c4a3c59480a89e567c013f3bd5d : FAILURE in 1h 29m 35s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/f82e8bbe2da94972b70d199db7ea7dfc : FAILURE in 59m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file trove/instance/views.py\n  Comments left for invalid file trove/tests/api/limits.py\n  Comments left for invalid file trove/limits/views.py","accounts_in_message":[],"_revision_number":3},{"id":"9207093b9df53b22296a8c19b1bacf7d06627ece","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-26 03:45:09.000000000","message":"Abandoned\n\nIncluded in https://review.opendev.org/#/c/730685/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"149ba37407a3d7cf34874ef2956274a387086772","revisions":{"67f8fca335d8a23e6497833b80e73e85a0fdca78":{"kind":"REWORK","_number":1,"created":"2020-05-18 12:00:28.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/728816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/728816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/728816/1"}}},"commit":{"parents":[{"commit":"a006dc41fb0925209f7cdb1d2547ddeee7510cbf","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a006dc41fb0925209f7cdb1d2547ddeee7510cbf"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-18 12:00:18.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-18 12:00:18.000000000","tz":720},"subject":"Use stable/ussuri for CI job and image build","message":"Use stable/ussuri for CI job and image build\n\nChange-Id: Ia3999c0cfbd104f723eabdf08c842e3f8d671b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67f8fca335d8a23e6497833b80e73e85a0fdca78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67f8fca335d8a23e6497833b80e73e85a0fdca78"}]},"branch":"refs/heads/stable/ussuri"},"e97f76ba99ffeb43d2e5fa8f73a15db1f997f770":{"kind":"REWORK","_number":2,"created":"2020-05-19 00:50:31.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/728816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/728816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/728816/2"}}},"commit":{"parents":[{"commit":"a006dc41fb0925209f7cdb1d2547ddeee7510cbf","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a006dc41fb0925209f7cdb1d2547ddeee7510cbf"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-18 12:00:18.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:50:13.000000000","tz":720},"subject":"Use stable/ussuri for CI job and image build","message":"Use stable/ussuri for CI job and image build\n\nChange-Id: Ia3999c0cfbd104f723eabdf08c842e3f8d671b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e97f76ba99ffeb43d2e5fa8f73a15db1f997f770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e97f76ba99ffeb43d2e5fa8f73a15db1f997f770"}]},"branch":"refs/heads/stable/ussuri"},"149ba37407a3d7cf34874ef2956274a387086772":{"kind":"REWORK","_number":3,"created":"2020-05-19 00:51:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/16/728816/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/728816/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/728816/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/728816/3"}}},"commit":{"parents":[{"commit":"a006dc41fb0925209f7cdb1d2547ddeee7510cbf","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a006dc41fb0925209f7cdb1d2547ddeee7510cbf"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-18 12:00:18.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:51:37.000000000","tz":720},"subject":"Use stable/ussuri for CI job and image build","message":"Use stable/ussuri for CI job and image build\n\nChange-Id: Ia3999c0cfbd104f723eabdf08c842e3f8d671b7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/149ba37407a3d7cf34874ef2956274a387086772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/149ba37407a3d7cf34874ef2956274a387086772"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
