)]}'
{"id":"openstack%2Ftrove~729075","triplet_id":"openstack%2Ftrove~master~I6d316779cc7220a855ce187437056b667bbe1f75","project":"openstack/trove","branch":"master","topic":"fix-ci","hashtags":[],"change_id":"I6d316779cc7220a855ce187437056b667bbe1f75","subject":"Fix CI job","status":"MERGED","created":"2020-05-19 00:28:31.000000000","updated":"2020-05-25 21:36:00.000000000","submitted":"2020-05-25 13:35:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729075-1590413735449-f8705a57","meta_rev_id":"4f18a2a258867a6ce5ca5d2ac2d0e1a99e34063b","_number":729075,"virtual_id_number":729075,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 13:35:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-25 09:02:44.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-25 09:02:44.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 13:35:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c537706afd3c6bb9ba59f4c52366a5ac351cdda8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-19 00:28:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad602833c91b06de25551aceaf234bbdbe453b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 02:18:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31285a8867f640e6b31a3838987d53fa : FAILURE in 6m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31b5ba1dd7704735809f4c37d43ef877 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f5d7549031d4dd8ae64e2978843709d : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/006316d13a564c879bd3e2266e164767 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51c71ec08017433abd85ddb3a8cf45a1 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f00230c30c4a2b8705838e7180b79d : SUCCESS in 5m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b925f1ad7cef4a40b8d3e42348190c74 : SUCCESS in 8m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/73c151062bf445c09ea2ba81145e2664 : SUCCESS in 6m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5c13795f46f4464c82b43bb966a3de99 : FAILURE in 31m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d66997f32a4c4b819eefc11314088931 : FAILURE in 32m 49s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6e58321bf6d2456eabe088369ca20c15 : FAILURE in 28m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/da2f638874f1496cba0d00f5b0fdb779 : FAILURE in 39m 00s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2471395a7ab2443da2f55400960d85bc : FAILURE in 32m 26s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/90d12cb4374940aca9e445af58557630 : FAILURE in 35m 51s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/9a401be4c54f45b0a12e90f01060bf88 : FAILURE in 37m 47s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/72ebe8256c5a409b8a1f0a2bc400ea38 : FAILURE in 34m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6636869837d3428d9addb908fe251391 : FAILURE in 32m 11s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/e004e03069134be9beab8fea7189b0e5 : SUCCESS in 1h 49m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b978727c1bfac24d3b7bf6980efdf4a9dde0db2d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-20 02:56:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"647b5737f290968a11d8863a5871fab513752446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 05:09:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec9a6adc0bda4abbbad03bef75eceeb6 : FAILURE in 7m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af8d5f30b3064b0ca6b925e363583be7 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b623bdbb3c74ddeb08a8394526ba055 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feccdfdb381f486b8141536091112139 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd4c360b8e044746905a946888320b35 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f7d89625e444108a24a59c6d63a34e : SUCCESS in 6m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35459540bcc24ae295d931ac9f30d923 : SUCCESS in 11m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57e03dd2e44244d8a17bf434bcfee1b6 : SUCCESS in 6m 31s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/63ad37613a8f4282bcbf7cb359d74590 : FAILURE in 37m 10s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/95f802880cf14a15a4dda4ec3b378d38 : FAILURE in 46m 00s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4e30c35ed2b345acbc3bed3881ad87b6 : FAILURE in 32m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/34f3074a2d8e4798baa42ac525b874ce : FAILURE in 36m 51s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/5ade5ca59b004f589f75f7448e3aecbb : FAILURE in 36m 25s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/8177be42b4024ab9abfefa43df6c79b7 : FAILURE in 48m 32s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ca12619d943b41b6807b67c7b92cb4ff : FAILURE in 37m 09s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/6127b48b5d4c4ae2ac4708de4fbd779c : FAILURE in 38m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/539897d00e59461b890353f6b02c8f21 : FAILURE in 46m 30s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/192355d74907485bb03fae29db1740c3 : SUCCESS in 2h 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db17970155769ee67e7c8e718c811505e0fabb1a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-20 22:39:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4fe754b8e9abaa8a79ea054b708acc032e774ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 00:35:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0778cc9ffc834231bb33510d5ba16c53 : FAILURE in 9m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/714052d161dc4515a9972e518a985fcb : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4724d4289884d96a1e5725a5db26647 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98b1c51de33643f48770244b02869478 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2fedbc0850c42a1bd1865aa3a26bff3 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184890e1b0a04d9eb8119be44d533ad2 : SUCCESS in 7m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c06ba092140b470487382b94bc2c09d6 : SUCCESS in 11m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f2b6415eea4405cb3d6d52ed7f778d6 : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4c18053c2fee4e8ab7d43ba7d2f7ba84 : FAILURE in 1h 36m 16s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7e03d20ba9e244ad97580ddfa8983f93 : FAILURE in 44m 05s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4384ae653de5445d823ac0b17d354009 : FAILURE in 1h 13m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5aeeb4556c794082b671bcbf21d8c8cd : FAILURE in 1h 14m 26s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/0afdec4c28054611b98ba854c13c2313 : FAILURE in 1h 10m 23s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/1b538d14594941e9b886dae8f8e38fd3 : FAILURE in 58m 02s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ea52c24e5b3248689ff4a5f7b599b60f : FAILURE in 45m 57s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c95c03cb986d4cff8bb3157706b8b958 : FAILURE in 1h 08m 45s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e77d13127344d78bd32549877841880 : FAILURE in 1h 36m 32s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/7de3169657bb48b2a1982fa5cc01bc65 : SUCCESS in 1h 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36862fa7c367128d5982b264d41c8822b16b36cb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-21 11:41:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d50b7189602168bea2de535a81d65ac48be0a48d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-21 12:27:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"829007ad55906d8b40ae0b38c9f8b37da73c8bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 14:26:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/550af333dd1e44f48804a744d0361d00 : FAILURE in 7m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/131ee4375bb045d699b45f38220502f4 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a98d9881d44264bed06ce625a81bd9 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5855c37eed0487e937965eb8293c9f3 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db893e5ce2c04393afea2365102e3bcf : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463a289a06b040498e1e123698ccfc63 : SUCCESS in 6m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c79a54bba53640d2bff914c2adc2f37c : SUCCESS in 12m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/195d9f779fa845afaf186c5c6576cbb8 : SUCCESS in 7m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c93748c8d2134e598f28e5288111ead9 : FAILURE in 35m 13s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ffd0a50c54f44676b9760ab7e1a13422 : FAILURE in 34m 38s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7a3646a6a6114f9d973d8f71967ecaf7 : FAILURE in 36m 49s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/4b2ba4740137454486294ae9a0d4973b : FAILURE in 36m 48s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/4fe808d1833e4510a5803bbfd826fbba : FAILURE in 1h 10m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/a9539caece36410a8dfbc2897690f7b4 : FAILURE in 44m 23s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/404d8be220754832bcc11f48d71843ba : FAILURE in 42m 54s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d612690a11344317aae962d1732a14e4 : FAILURE in 1h 02m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01c0136b23bc49d4ad559b6aa3901b66 : FAILURE in 34m 46s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/e6f9406bd75a4a9aa9aa6585619544dc : SUCCESS in 1h 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"58c31f4379812ad1679a5e6da6562016a0c0e5f9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-22 01:39:03.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"24ce39992cf365a10375b3b522b340fdf7aab771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 03:34:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2491f8b0a93f47ddb7e1384f65696d88 : FAILURE in 7m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/746f5204e3144433ab7de1bec7e7b152 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315857702d1848159f9e96789f4e0e63 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c88e8c47b55249eb8e5336960d27f694 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e78fca7523c24d918043f7d4c699f94d : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04bf269315494b0ca64f6d4c1b4ebf08 : SUCCESS in 5m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea8d890ca5294bf28cf9be6397484806 : SUCCESS in 8m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9f30a7593c2642a48c17b0698401c18d : SUCCESS in 6m 29s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/41e827f3209c4461bc7ed53744c40f9d : FAILURE in 1h 33m 05s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/a88af40dc8ca477aae34b4082ed83d34 : FAILURE in 57m 00s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/3226ff505f81469bae06618761adf0a7 : FAILURE in 1h 10m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/75cd43dddb5a4d74bd7712eea490c1e7 : FAILURE in 1h 04m 32s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/37bec37d700049138f9d274bafbfda9d : FAILURE in 1h 02m 59s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/c2107f4f974e4dfd899bb1ef3a181537 : FAILURE in 40m 02s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5a7dd0d6fa50408aa49db8057a3d00c8 : FAILURE in 33m 24s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b05cde40474f49d493fd0fabd45d794d : FAILURE in 1h 01m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca091e20787e4cd286ad2632052581c4 : FAILURE in 30m 21s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/6f0086b2e18c40efadb490c38147e7c3 : SUCCESS in 1h 54m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d1c136cfeca859442842ecff4e7d0877077aef07","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-22 05:40:20.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5bac22d76b41119da65d3276752fc8ef6ef938c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 07:48:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8678bf05749340ef90ae164575430e96 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/318ebf6656644bad8c6eed418ea86fcb : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f6c19eacdc4dcba0e9063e0bcfa877 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61960e551f9d4002a8f68bc4d559da19 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15bfbf747ecf46c7b853b35bf5252d02 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb0858307d1541f29a60fb8f19c41fa5 : SUCCESS in 5m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36f9a37c065d4ad9bdf2f799bb933f82 : SUCCESS in 8m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/deea5220300c40be847d8ec192a43d48 : SUCCESS in 5m 31s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ec4b00642ada441bb8a09c5e1533733f : FAILURE in 30m 47s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6738e79b4a8743cfa946c1a7165af314 : FAILURE in 59m 50s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ba800995f150466093fa58d066a1e449 : FAILURE in 1h 18m 23s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/77bd603e38854d76bcef10c896826bd3 : FAILURE in 33m 21s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e1d09a0d1ab2478f957b921e6e68ad39 : FAILURE in 1h 08m 22s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/9f9b3f991fc64a059afac310bccbdfa8 : FAILURE in 42m 37s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/35bb5b4bbec34cc4ab8cb365f09afcf2 : FAILURE in 42m 26s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/752b421210e041dbb75f667fb3ba3429 : FAILURE in 1h 04m 26s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a20b33522dd40b282b127aacdf9ee4c : FAILURE in 1h 32m 49s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/3c715ad0a50b42a6ba2152ac1dddbba7 : SUCCESS in 2h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3a2fb667dda8c76e95efc845bfa10015e60d213d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-23 11:09:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ad1b19d9ba11d6e67fbc2f4a036ce65311e3595a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 13:47:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47ea8999585f418ea0f9f44c5fef19f7 : FAILURE in 7m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62850de6a54a4aa6b05f02a8777f5bcd : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4498f513684dd99d26a317acfd410a : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b6c2fc456e428f87928684905e6244 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e9b9e8d5cfe45eca9cc8718181651ca : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a402ca709db4fa3b7f6053aa1e59bde : SUCCESS in 5m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95c58b686e294ed5b609c62b73914ece : SUCCESS in 9m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/398d599d56724ef08d3251ba0eb03d37 : SUCCESS in 5m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6cab8ec69a954d2592407f0e6b0a7f6f : SUCCESS in 45m 38s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/55fb4178affd45e189ccde7678cc75dd : SUCCESS in 2h 07m 51s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/2a1c4f4966d44acf9821fdc3deb7fca2 : FAILURE in 57m 10s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/089f01855aac417b8c114bce395621c7 : FAILURE in 41m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/562c867c0fe14ae2a0ec7d8d8c4d0013 : SUCCESS in 2h 36m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/6773225594fc41a08d6ac10df1ee47df : FAILURE in 44m 04s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/00b053c5ca334212b4998bd0a8d10345 : FAILURE in 39m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/0125e269396644ad9797d4051ad64264 : FAILURE in 1h 18m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59e0e5cbfa784bcb8e7876deb0058533 : FAILURE in 1h 38m 09s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/1301ec3ffe1e47b4b11e8b0098e3bb1c : FAILURE in 53m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"52e1af49bac67a7371de388ed2a61c37e72bcbba","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-23 18:06:16.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a86ecb1b9ea711ef7adb9b70889b8ca700541c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 20:21:36.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/349aae3b8bc64bc39394d235b87974da : FAILURE in 6m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d146fa61e4104fd7931c657095cfda47 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa9522a89374d7c9aa6e398923601c1 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/203e4720a98c41958c7785fd03eecab2 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5acf41ba2759485fac6b4684eeb1b2f9 : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166897b781fb46bf8b0e500385000bb1 : SUCCESS in 6m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa9b30a7d01d40bfa054aca89e538acd : SUCCESS in 8m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a3605151e3c9494bbc145fbefc933dbb : SUCCESS in 6m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/06f19afd39d14b109dffbee8abf4175b : SUCCESS in 54m 17s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ae29037ac49f4331b1924b0b32ef32c9 : FAILURE in 37m 20s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c9c6e7d2802049ec9ca3a5677b6af046 : SUCCESS in 2h 14m 18s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/261ea0f20f284ae29116b3fd2439cacd : FAILURE in 24m 43s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/577c4a72e2144e2cb2496bb561157697 : SUCCESS in 2h 01m 35s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/460c793d8c384c5892f9939da5f5e12b : FAILURE in 39m 57s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/754c7507739843a890fa339af259e9d5 : FAILURE in 42m 51s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/2a780df02e6547d9a0ab42f16d9a444a : FAILURE in 1h 08m 37s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/282a61d734f04914893ef3e68bb43adc : FAILURE in 1h 09m 29s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/d88a4807a3ff4b90b47ed804993bb7b5 : SUCCESS in 1h 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce6b04f6916b9cb249649684855dcdf0241740f1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-24 04:48:57.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"65a4abefe4e979e299fa7a0f85213043feb1a719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 07:23:06.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d004841a7be49d59850c2bd002e07ea : FAILURE in 5m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ee0e58db83945779ae4d3a252d7174f : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4782f7603be648b3bb12a3316c212269 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8890fb500cb41cdbcdf33b61e5a53f7 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66bdf8174402428fab6ba3129f1989de : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecea865afc024ffbaf407b8261bcd6d5 : SUCCESS in 5m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8658d2a030ad4e78b54b85787f98f588 : SUCCESS in 8m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6ce2b51667942508dc4ce5f0aad9b7a : SUCCESS in 5m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2a5deb695c0f463f9c371bdadc0f3951 : FAILURE in 1h 16m 36s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3bee2142f0a5470f99181d99b8674a79 : SUCCESS in 2h 18m 32s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f88c623d4083408c9cbb40f609e9fc57 : SUCCESS in 2h 28m 50s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/16a1c9d08e2d403998876ed083871b94 : SUCCESS in 1h 29m 44s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d060f51a883441eb9e2b8f0f03fcc8cb : SUCCESS in 2h 14m 10s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7b9e1a28cb9642f781a5da5d39c63a50 : FAILURE in 48m 00s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/685cb05becd0476a8521daae621ca388 : FAILURE in 47m 59s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f9851008c56d4293b599aee4947c54ce : FAILURE in 1h 13m 19s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2d05f418024fadbdac600decc12140 : SUCCESS in 58m 38s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/28f8d3db2ee943f0bcb9e5c65f1493ba : SUCCESS in 2h 09m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"42012b44ea5c4b5b05dc1758a4e06adfdb770a20","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-24 09:44:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ecb93851375986acf5b195bac6ff2f82715928e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 12:05:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21d74c8719cf429ab5eb37738624dec0 : FAILURE in 9m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23083ccad7f3422eaa74b29e65988e9d : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36f0f4e5c6e463baf3df4b7512d0e1e : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd80d65673894a76842d9fdc132006d3 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/573d443f6e6a4110905fd1623b8f04ed : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0c8595a6246498a90d7585c780586c0 : SUCCESS in 6m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/41f78cec299549898f920d0ba4606f2a : SUCCESS in 8m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36fde2ec103f4252af99400543593775 : SUCCESS in 5m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/cc78e247607c47eeabb17cb183eec88b : SUCCESS in 1h 02m 45s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e2997f7201c04ac099196ff54185153e : SUCCESS in 2h 17m 48s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5b8ac64cdd8d494ca3753d2bcda11785 : FAILURE in 41m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/1e5b10df5c204438b7c3f0d2e14dc113 : FAILURE in 40m 13s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/bfb5af5da4264a729a8374e08ed36ab4 : SUCCESS in 2h 20m 08s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/720261e18633436cb99eed793c860182 : FAILURE in 43m 29s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/bd18ea4e61c14286adaabf0a3b9b62ef : FAILURE in 46m 25s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c01d045c7b46442f972b157d7529ce81 : FAILURE in 1h 12m 22s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413112987cde4a62b293d79de60bfede : SUCCESS in 1h 02m 08s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/da1b63639a8b4bb8b837574b7e8f9910 : SUCCESS in 2h 04m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"41bf6495c37e680940b9f550ef8e174d5cffdbd9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-24 20:43:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"96ad0433b3f1614c43d62ce16f4c5d70574bf91c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 23:15:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbee9b3edaad4e2498edbe622519757d : FAILURE in 9m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47cf05a9884340b888d5219e469bd998 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f7a09c7ff047e9b005851bbbc9f6ed : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f16469e055e14613b875664e2a61705b : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b6ca7b25b54daa9364be7d7838bc54 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79952ea91e0a4730b7b488079bce1e22 : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cb0e71e331b4b65a72b673f159815b4 : SUCCESS in 9m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b91ee128d9094faba6ff0b39f27b4b5a : SUCCESS in 6m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/17ca7237339c4c4d9b783aa14ac355fa : FAILURE in 40m 35s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/40827c0287a2438b93a4e147c323730a : SUCCESS in 1h 47m 58s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/757215733427466988639e9c07a51b67 : SUCCESS in 2h 30m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2eece38caf08424eb8534cb4e0dccd5f : SUCCESS in 1h 37m 31s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/540a30b9fb3a4912a2f40f41a6cd0ec5 : SUCCESS in 2h 14m 52s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4efc9fe707234a6eb451ab50220b3060 : FAILURE in 40m 56s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ac9cfef1d30748ad9f4bf591027bd320 : FAILURE in 42m 27s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/2f401287b86047ada7fbc7b80022cbfe : FAILURE in 1h 04m 33s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb960b325eba4096ab542ea782319aad : FAILURE in 1h 30m 11s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/6da5501111344e0a90cb74d0c9ad9f28 : SUCCESS in 2h 05m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c51909073397a513ff0a84ee1d68f6ea4389ec9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 23:15:15.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2d50741f799c6ddd0f2242613d0c5fdd8fda9794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 01:04:00.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c90fab3a7bf941e6bd2a54e1d909ca03 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1090d6eb669d486da30352bbefd79db0 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/646927b2ca724c7ca3a543f5b947b12c : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5983f955f804fd180532b9a1979a11c : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fa679bca1f41518e56eba594f67646 : SUCCESS in 5m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c6c53df3ba2a4e10a20579e54591c61d : SUCCESS in 7m 54s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/674fee36a5b746b1a7453731427324b9 : FAILURE in 26m 24s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/6af233bcbcea4cd49d2520367609ecbd : FAILURE in 30m 12s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/37e4ca08a60b4b71a16db3f571fdf456 : SUCCESS in 1h 47m 49s","accounts_in_message":[],"_revision_number":7},{"id":"ba2af733b599e841715040384571592bc778b9ad","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-25 01:27:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4767da114c1a80513720d5e6b8152d83f99a120b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 03:38:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9dc08a4ddfa4dd8a7060079157a99cf : FAILURE in 7m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13a6b7833e014db4b890d86dd604eb49 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81576ffeea7945bdb254aad7b2b148f6 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/305b670658bc4e9187ed6d7a3b716cd6 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b67827dfd3754fa1a6a7c4f968ce5eeb : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed33cc2ee104d67840fe735425078dd : SUCCESS in 6m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd421e6bab4f4d04996806f8498474df : SUCCESS in 9m 16s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dbf0ee9ea924f2a8ba7cc907266b38f : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/378fb456703c4805b9038d3b8483d92f : FAILURE in 30m 52s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f31e43300f094dcd96fa0267f6d0523d : FAILURE in 30m 11s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/4dee450f985246aabb9919e7d97866d9 : FAILURE in 31m 07s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5c904e88214f40efac82b664ed28cd81 : FAILURE in 27m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/83e9db91d6354a2d8021c624857974d9 : SUCCESS in 2h 09m 13s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/90b205ee59e945739a903838810755c5 : FAILURE in 41m 20s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/34ab02c294c74beaba1c4ef686628a77 : FAILURE in 36m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/1dd0ea4458524f93bff5d3764e14ad83 : FAILURE in 1h 00m 52s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a36ad812391c41a09508a0bb62d3e2ad : SUCCESS in 59m 13s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/68d826052d2d4e269a39db8da34372bc : FAILURE in 1h 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"53607c99f2c471dca6cc0307572aea4689d425d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-25 05:37:49.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2e9e6cfaa5b6eb692e1185cae84a87725cba3d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 07:49:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e90e8b8454d649bb9015dc90eafc9ebc : FAILURE in 7m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d01f51d2dda1493faeacb9a9fd78d981 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43cfb6d7e744efbb093835786741e12 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/328bedf5cb9d473dabc388da061ca522 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bb9b628e0cf460185b70450535ab3eb : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc72426873eb4bdc92c811b74bf201a7 : SUCCESS in 7m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3a6c382f0c634b398d92a3e7b2e7a4b9 : SUCCESS in 8m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3dcecd7505f24ccd8b03afd97bc412ce : SUCCESS in 6m 12s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7ccefa3441004b7ea19a5854847ee395 : SUCCESS in 55m 58s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c915420d486b4b818d0098a35a4e0865 : SUCCESS in 1h 59m 57s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/7324d060b2d248389b01d9207df13aa9 : FAILURE in 2h 08m 15s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a005b8f9b7b34c7bace0d65443338f7e : SUCCESS in 1h 35m 08s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/084f614e702f4cfc802c37fe46b5a9af : SUCCESS in 2h 02m 57s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/4153780a5ec94edbab6543e8ee9ff311 : FAILURE in 42m 51s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/89d7b87f94c44e7f93d79e76f116e595 : FAILURE in 43m 27s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/18d62b945b8f4765bfbadb656cf7e4f9 : FAILURE in 1h 04m 35s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a071a75e5fe456493d1be16f54355ba : FAILURE in 1h 11m 55s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/53c5760aefce4a7b9308cfdd8e236693 : SUCCESS in 1h 49m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2d6f20ddb16b5752f47d69788948b5019b0dbddc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-25 09:02:44.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e16a85d5b64e4b1f84f64fafa9a307c280f81c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 11:35:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7167480ca247638e759ce1326c0d45 : FAILURE in 7m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a75a70724c354b4e92d01a5fc2a8a411 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8ec4063eb34f29beb01c1999744c5e : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee35d1313eea46cfb65bed16a8fdab51 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cba59fc9c7b741c5873df7c44ae639d9 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f76970ce2e34ad199a887b20488abd7 : SUCCESS in 7m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0676656622d4412a9d4b3528f75339e : SUCCESS in 10m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3374b7cf9b1f41c297566f1e14355f44 : SUCCESS in 7m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bbd17e340b3e47f19aec6397e77ea78e : FAILURE in 1h 19m 46s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/40dd0c909d314699a200b6bd41a67c51 : SUCCESS in 2h 05m 18s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/33a33ff9dcdb488f88c7f7c28bef691a : SUCCESS in 2h 27m 12s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e5c107693c89474ea5b7b8fd2959b0ea : SUCCESS in 1h 35m 28s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/a033f280831641e9b7adc3c39a03eebc : SUCCESS in 2h 11m 29s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/30e781a2e7234e24a5f975a88c43df4a : FAILURE in 43m 24s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/46fb57cfb5c44dcd877608146adae29b : FAILURE in 47m 54s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/4d4ca84b245e4a8d8f24ca69b543678b : FAILURE in 1h 14m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a8f172302b4e03a1e7858668ea64f6 : SUCCESS in 1h 00m 21s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/46d8088ea6e0412296700440bf555938 : SUCCESS in 2h 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7f5cb99d920d76fd9f496dd1abeb1ad43de9faf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 11:35:36.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"07f5715a07e35995ab59b8ee77b39cbf2536176e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:35:35.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779f73dfc2744fe782c110905d22b775 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e06beca38a4b9dbcc22f39a78ba672 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98e3598ae52446c890e81f5bcca52b86 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/634cfeddc10f428f92ab2947878392d1 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7bd53c2df84dc59bfc4e16f04e5df6 : SUCCESS in 6m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0954f8ebb8214a55a1b31eacc4a4bcd3 : SUCCESS in 9m 41s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/5ac3d69629284798a2afe035787dd7fc : SUCCESS in 1h 55m 39s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d65e8c39ec0d4dc78a076781f8f9b175 : SUCCESS in 1h 58m 26s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/dc6d22ed78aa408babdb0aca939d86ca : SUCCESS in 1h 55m 27s","accounts_in_message":[],"_revision_number":9},{"id":"601418118712a59bfa4f3401652a599143410965","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:35:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"290bbaef44b1b04a806961d905a0bfbb986dc09d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:38:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96defeb214a74e87b56fb13423279786 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":9},{"id":"8fcf43cf4e5de4ca29b4c41d33300210cf00fca3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-25 21:36:00.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 76f738c7999faf013bfed6c0e6f82bba1e0d8b17","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"523d66e8fd5d29be8dbae9aa79c5348d3dce8c64","revisions":{"3c396593bac3169c4ddbf972cb123d20ea67f086":{"kind":"REWORK","_number":1,"created":"2020-05-19 00:28:31.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/1"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"subject":"Fix pep8: ambiguous variable name","message":"Fix pep8: ambiguous variable name\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3c396593bac3169c4ddbf972cb123d20ea67f086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3c396593bac3169c4ddbf972cb123d20ea67f086"}]},"branch":"refs/heads/master"},"61f9166bc0e112e61d1fed9d7fe02d15bffc0b49":{"kind":"REWORK","_number":2,"created":"2020-05-20 02:56:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/2"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-20 02:53:20.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f9166bc0e112e61d1fed9d7fe02d15bffc0b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61f9166bc0e112e61d1fed9d7fe02d15bffc0b49"}]},"branch":"refs/heads/master"},"76d316d1365718261a1f1276a229dfa857265be9":{"kind":"REWORK","_number":3,"created":"2020-05-20 22:39:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/3"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-20 22:38:58.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76d316d1365718261a1f1276a229dfa857265be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76d316d1365718261a1f1276a229dfa857265be9"}]},"branch":"refs/heads/master"},"78954c859234debfdffc020a7673a1c19fc8aadd":{"kind":"REWORK","_number":4,"created":"2020-05-21 12:27:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/4"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-21 12:27:09.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78954c859234debfdffc020a7673a1c19fc8aadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78954c859234debfdffc020a7673a1c19fc8aadd"}]},"branch":"refs/heads/master"},"c5444c192ec3d7520afbc5a570df169a49f6b629":{"kind":"REWORK","_number":5,"created":"2020-05-22 01:39:03.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/5"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-22 01:39:01.000000000","tz":0},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5444c192ec3d7520afbc5a570df169a49f6b629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5444c192ec3d7520afbc5a570df169a49f6b629"}]},"branch":"refs/heads/master"},"69b5b25e60fd43f5534c72b071d1d54ba1f6d69e":{"kind":"REWORK","_number":6,"created":"2020-05-23 11:09:47.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/6"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-23 11:09:32.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69b5b25e60fd43f5534c72b071d1d54ba1f6d69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69b5b25e60fd43f5534c72b071d1d54ba1f6d69e"}]},"branch":"refs/heads/master"},"88fa7fd307d3f3fb33ef2f0f14a3278b8a534515":{"kind":"REWORK","_number":7,"created":"2020-05-24 09:44:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/7"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-24 09:43:24.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n* Mark tempest job non-voting temporarily because of some tempest bugs.\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88fa7fd307d3f3fb33ef2f0f14a3278b8a534515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/88fa7fd307d3f3fb33ef2f0f14a3278b8a534515"}]},"branch":"refs/heads/master"},"0441f320dd86e120b322330332046f198bc34fff":{"kind":"REWORK","_number":8,"created":"2020-05-25 01:27:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/8"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-25 01:27:17.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n* Mark tempest job non-voting temporarily because of some tempest bugs.\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0441f320dd86e120b322330332046f198bc34fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0441f320dd86e120b322330332046f198bc34fff"}]},"branch":"refs/heads/master"},"523d66e8fd5d29be8dbae9aa79c5348d3dce8c64":{"kind":"REWORK","_number":9,"created":"2020-05-25 05:37:49.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/75/729075/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/729075/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/729075/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/729075/9"}}},"commit":{"parents":[{"commit":"ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268","subject":"Add command example for preparing datastore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae4d1a5ebff2b8de8dac4dcd5be000c8f1bdf268"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-19 00:28:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-25 05:37:38.000000000","tz":720},"subject":"Fix CI job","message":"Fix CI job\n\n* Fix pep8: ambiguous variable name\n* Install Python 3.7 for trove guest image. oslo.concurrency requires\n  python 3.6 or newer. See\n  https://bugs.launchpad.net/tripleo/+bug/1861803\n* Mark tempest job non-voting temporarily because of some tempest bugs.\n\nChange-Id: I6d316779cc7220a855ce187437056b667bbe1f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/523d66e8fd5d29be8dbae9aa79c5348d3dce8c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/523d66e8fd5d29be8dbae9aa79c5348d3dce8c64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
