)]}'
{"id":"openstack%2Ftrove~730979","triplet_id":"openstack%2Ftrove~master~I511faa17b27eac827922957e4b36108da0dcd0c8","project":"openstack/trove","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"I511faa17b27eac827922957e4b36108da0dcd0c8","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 20:14:39.000000000","updated":"2020-05-27 05:07:25.000000000","submitted":"2020-05-27 05:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730979-1590555961208-65991141","meta_rev_id":"201e353987abb791784e277b8164118fcb8213f2","_number":730979,"virtual_id_number":730979,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 05:06:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-27 01:56:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-27 01:56:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 01:56:35.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-05-27 05:06:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd39b9067b1af4d8d9c22ec489f5a9be797010e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 20:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe8afc056a060be59aae598397d0a20d26798a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:55:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e449b147a6dc42bb8faff681742c6ea2 : SUCCESS in 6m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d437a22390fa48bbb46d39a0a4d9ece2 : FAILURE in 8m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc11e9af8f064c71a1bed85c9bb11941 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2626bddc8db34f4f8878d45b24cc31da : SUCCESS in 9m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b292963d58ae48b6ac511ee56d8a5e66 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9690c8610424912b1018accae68508c : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87dc8e64b9ee4cdbb02bdf3387bb9e1a : SUCCESS in 10m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bef7fee9b90c48019cb0b369c65e7775 : SUCCESS in 13m 22s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/58ce7d28d8534cc4b766a79535c20631 : SUCCESS in 1h 17m 44s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fb99a489580f44f08872024c0be52d53 : SUCCESS in 2h 16m 25s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/f3527a75407e4d01880c1411add77076 : SUCCESS in 2h 17m 57s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0397643b7c5a4ff58f7ac02983b053b9 : SUCCESS in 2h 02m 08s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fd0081bba36f46e98cf6a21e98de55c1 : SUCCESS in 2h 04m 21s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3a0a1c631e77486b8624e063430ddb0c : FAILURE in 48m 43s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/45820cd2b4f9465099890cc2e11f2e7f : FAILURE in 43m 13s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/fd32b230e6534181a4db3ca699d423cd : FAILURE in 1h 23m 13s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45d1ad349a7c4c209b5c665ab8ef5eff : SUCCESS in 1h 06m 29s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/f49922c1ce4d45e4888a33cbe053bb4d : SUCCESS in 2h 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eeb7031cb83b94ba1bd075578e0a3127b850b986","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 01:56:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d921c2ec19f284ab8725f44e2f3d0b9072455b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 01:56:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca820912cf8e0c6eeaccde2012af0f1cf2246909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 05:06:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/862259c93b124231b7909b4c2cf8ed6e : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c73dc51ce02a4249b7851f02ad3d4144 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ef1a74d3dee40c2846558c3fe0cb643 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c462697f77714c819ce9aa4918bcead3 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93233dfc1f43468bbdf95f02b0020daf : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5be39163d547a5ae52d20e83f43a72 : SUCCESS in 5m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a36b839e05904cec83cc3432d1f6e3f7 : SUCCESS in 8m 12s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e95f747f83824b169bc0b894afc855d3 : TIMED_OUT in 3h 03m 34s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/fa32578c02a34f68a03c5564ef607222 : TIMED_OUT in 3h 03m 44s (non-voting)\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/3a5a59dcc4264189bc69481e904611a0 : FAILURE in 2h 55m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"088bf32dd2ede27781aa2d3871adf54bd0cb2e8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 05:06:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e4e46c5bcff54301e979f74e8bf9a241d1dc3b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 05:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb4463baba1a483693a401aadee70c8e : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed726fd2f7da022b25e342554e782682f3581416","revisions":{"ed726fd2f7da022b25e342554e782682f3581416":{"kind":"REWORK","_number":1,"created":"2020-05-26 20:14:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/79/730979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/79/730979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/79/730979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/79/730979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/79/730979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/79/730979/1"}}},"commit":{"parents":[{"commit":"523d66e8fd5d29be8dbae9aa79c5348d3dce8c64","subject":"Fix CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/523d66e8fd5d29be8dbae9aa79c5348d3dce8c64"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 20:14:31.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 20:14:31.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I511faa17b27eac827922957e4b36108da0dcd0c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed726fd2f7da022b25e342554e782682f3581416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed726fd2f7da022b25e342554e782682f3581416"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
