)]}'
{"id":"openstack%2Ftrove~731168","triplet_id":"openstack%2Ftrove~master~I49bd6c824b837d22d9f3945a3e8c2343c831b352","project":"openstack/trove","branch":"master","topic":"doc","hashtags":[],"change_id":"I49bd6c824b837d22d9f3945a3e8c2343c831b352","subject":"Trove doc and CI","status":"MERGED","created":"2020-05-27 11:34:44.000000000","updated":"2020-05-29 01:04:28.000000000","submitted":"2020-05-29 01:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731168-1590714036068-65991141","meta_rev_id":"b149a71dbe4e65d5d8c4c44794fdcd991ee71e22","_number":731168,"virtual_id_number":731168,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 01:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 22:12:44.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 22:12:44.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 01:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1267371a70d464b39c81814081e1a9bdd0c3a1f5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 11:34:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7eed29dff7504c94b354a7fbe92c1937afaaa38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 11:43:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c3c3b55a004eb8bd86f38654e3949e : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d2769f1d8f4cae967d827d6f276d37 : FAILURE in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13d370ad394e46b7891228f62254ffb7 : SUCCESS in 7m 02s\n\nWarning:\n  Comments left for invalid file doc/source/admin/index.rst","accounts_in_message":[],"_revision_number":1},{"id":"c450021db2f46e2c2ea2cdf0169703250fdb894b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-27 22:18:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5aa46e028259e4d3baabd0934368e67a6896598b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 01:41:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55335559b6494b71a83ff009f6eb0a82 : FAILURE in 9m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45a0b66f4a9d4c5ab5e483cae158a0da : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29a6b3ab8614c42b3feb77a6df4f165 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b64e940352640b6b38800b132b87f5c : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10fde2f53d784521adcf90cd9b20abbd : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d354ba9ef104d94a8a40275be5a530f : FAILURE in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf60cd1ed06b4c0f8728474f61d7a920 : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/caf6a68728704806b0c3c2659ca57a31 : SUCCESS in 8m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b0d7b512e7a4a5486bf929700c1edfc : SUCCESS in 6m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ccd2733257794b2180345447dfe20a6b : SUCCESS in 1h 14m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b9118ac572f04abd9aa2c098c7fde070 : TIMED_OUT in 3h 02m 30s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/72a9d8a2817e40c28c1ddbd99201786d : TIMED_OUT in 3h 02m 58s (non-voting)\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/92eeea2c75a24605ae1b379dd5d404bd : FAILURE in 2h 28m 53s (non-voting)\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/439aa54273cb444fb7cce5a5e407b536 : FAILURE in 41m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f403e9eb817e46128ce745133e58856e : FAILURE in 39m 20s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f420455b0ba4d4f94bd3dc32cc92cb0 : SUCCESS in 1h 19m 27s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/b1ccb1c2d0164471a54eed01db1d8fc1 : FAILURE in 1h 09m 28s (non-voting)\n\nWarning:\n  Comments left for invalid file doc/source/admin/index.rst","accounts_in_message":[],"_revision_number":2},{"id":"d1fb4482703168eebc944cde85bb74cd36e28d0c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-28 10:36:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"008e519c3111744a2d696e53646e492592bf8d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:36:29.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  The job \"trove-devstack-base\" exceeds tenant max-job-timeout 10800.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: trove-devstack-base\n      parent: devstack\n      abstract: true\n      timeout: 14400\n      required-projects:\n        - openstack/tempest\n        - openstack/diskimage-builder\n        - openstack/tripleo-image-elements\n        - openstack/trove\n        - openstack/trove-dashboard\n      vars:\n        devstack_services:\n          tempest: true\n          tls-proxy: false\n        devstack_plugins:\n          trove: https://opendev.org/openstack/trove\n        # NOTE(zhaochao): we have to override the default settings from the\n        # the base devstack job about \"SERVICE_HOST\" and \"HOST_IP\", as trove\n        # guestagent should use public addresses to access the underlying\n        # components(MQ and Swift), and the previous devstack-gate jobs didn\u0027t\n        # set these variables. However we cannot unset variables in job\n        # definition, so we just set SERVICE_HOST and HOST_IP to empty strings,\n        # and VNCSERVER_PROXYCLIENT_ADDRESS to 127.0.0.1 instead.\n        devstack_localrc:\n          SERVICE_HOST: \u0027\u0027\n          HOST_IP: \u0027\u0027\n          VNCSERVER_PROXYCLIENT_ADDRESS: 127.0.0.1\n          USE_PYTHON3: True\n      run: playbooks/trove-devstack-base.yaml\n      irrelevant-files:\n        - ^.*\\.rst$\n        - ^api-ref/.*$\n        - ^doc/.*$\n        - ^releasenotes/.*$\n        - ^tools/.*$\n        - ^trove/hacking/.*$\n        - ^trove/tests/unittests/.*$\n\n  in \"openstack/trove/.zuul.yaml@master\", line 61, column 3","accounts_in_message":[],"_revision_number":3},{"id":"910eaa29eb8906a7c82ab5a27c05fee50d6dd103","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-28 10:53:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6374efd48d3283936ba1c31e4d2cd739d0e16514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:17:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/071f831cc84443aaa47de1f926f99e0f : FAILURE in 8m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea03e47202af4a859146f260334e1a29 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d5e09895f54022be79447ef8f76d36 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2768df2432d44848db5c47366795f11 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5ee67d5ed74646a2f90905bd496afa : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6160dfb085b458287ed239f8e5007bc : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20157cabe8054dfc86766d4e3ee66323 : SUCCESS in 4m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64beab30a3e845d690bbc85219f1fb9e : SUCCESS in 8m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a6d1fc18a834df38846cdd47612e3e8 : SUCCESS in 7m 44s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c578dfa95470414c974ffba700a4e5fb : POST_FAILURE in 1h 19m 38s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f99f6986002840748c659bc0b0c5e16c : SUCCESS in 1h 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc56aaa031539e00d7d77ecc38e67e6e160a7833","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-28 22:12:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"517b241ff4be85724c32ecf38870f459223d201b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 23:38:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b8a0239ad654a9c9ea4f8ce5c56bc09 : FAILURE in 10m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d97351366a9e47348a52874054c5c1a9 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51885bbe521a40229ba4c0c4b94360b7 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d6117e64d034ef8b7efca05f926d784 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57587c56f1a14ddabea87d559dba877a : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be0b69456934fb696fb25134bd271b9 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e62cd0b059c467f9fd78d08cbc11389 : SUCCESS in 5m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d3854ad2fbb7407bb878c2c9dab2f197 : SUCCESS in 8m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b86880fe568448c8929a7973ae4fa656 : SUCCESS in 7m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2c09edb9085b4340865a8fd650eb3735 : SUCCESS in 1h 07m 49s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d235baa659de4a5987dde44ca759191e : FAILURE in 1h 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f45968f5a766c025a1dbca329b88de1d6ea34855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 23:38:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fd86ed8b7087d58f7b70d9d21626f7a9fa7d7ef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:00:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e08ef8c765c4ba58f682abe670423c8 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09136fedd8ab445bae9ba5fd15df266e : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45f7e3c69449431a8fc2a2ccd0c806b5 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0744dbc43f594aa09fb0098f81dbb9c3 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f846e4be19334cf290ef398ef09bdd90 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8c027da2648417d8dc102b8d03fb20a : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/afb10030554345e084fb4d5ccce8e787 : SUCCESS in 8m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4d834a74a07d4032bd9226e2113d6c57 : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":4},{"id":"615878b51dbdd2291cafc1b39b4c7b1f8847477f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:00:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"eb50bdece67f61230b96d72b5138acd7fe36c3b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:04:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d7e8ababa846039757165127499bf8 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cbe8e69874e4c53b38f2cb0e80a7637 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e2b26fdd7d60b82938682f9c3f82163da9b629cf","revisions":{"7d8464eb0e292a84ea86561a4b4d7afab964ff3e":{"kind":"REWORK","_number":1,"created":"2020-05-27 11:34:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/731168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/731168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/731168/1"}}},"commit":{"parents":[{"commit":"91a6e5fa8aebfd579271a2f78536d526142d6228","subject":"Merge \"Update docs upgrade datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a6e5fa8aebfd579271a2f78536d526142d6228"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 11:34:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 11:34:34.000000000","tz":720},"subject":"Fix Trove doc","message":"Fix Trove doc\n\nChange-Id: I49bd6c824b837d22d9f3945a3e8c2343c831b352\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d8464eb0e292a84ea86561a4b4d7afab964ff3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d8464eb0e292a84ea86561a4b4d7afab964ff3e"}]},"branch":"refs/heads/master"},"3ece21ac2035029e283bde414679fa827ac04e40":{"kind":"REWORK","_number":2,"created":"2020-05-27 22:18:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/731168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/731168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/731168/2"}}},"commit":{"parents":[{"commit":"91a6e5fa8aebfd579271a2f78536d526142d6228","subject":"Merge \"Update docs upgrade datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a6e5fa8aebfd579271a2f78536d526142d6228"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 11:34:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 22:18:47.000000000","tz":720},"subject":"Fix Trove doc","message":"Fix Trove doc\n\nChange-Id: I49bd6c824b837d22d9f3945a3e8c2343c831b352\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ece21ac2035029e283bde414679fa827ac04e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ece21ac2035029e283bde414679fa827ac04e40"}]},"branch":"refs/heads/master"},"180bd278febb27b2d2077644d60e087647eea06a":{"kind":"REWORK","_number":3,"created":"2020-05-28 10:36:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/731168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/731168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/731168/3"}}},"commit":{"parents":[{"commit":"91a6e5fa8aebfd579271a2f78536d526142d6228","subject":"Merge \"Update docs upgrade datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a6e5fa8aebfd579271a2f78536d526142d6228"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 11:34:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-28 10:35:53.000000000","tz":720},"subject":"Trove doc and CI","message":"Trove doc and CI\n\nChange-Id: I49bd6c824b837d22d9f3945a3e8c2343c831b352\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/180bd278febb27b2d2077644d60e087647eea06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/180bd278febb27b2d2077644d60e087647eea06a"}]},"branch":"refs/heads/master"},"e2b26fdd7d60b82938682f9c3f82163da9b629cf":{"kind":"REWORK","_number":4,"created":"2020-05-28 10:53:26.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/731168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/731168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/731168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/731168/4"}}},"commit":{"parents":[{"commit":"91a6e5fa8aebfd579271a2f78536d526142d6228","subject":"Merge \"Update docs upgrade datastore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/91a6e5fa8aebfd579271a2f78536d526142d6228"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-27 11:34:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-28 10:53:17.000000000","tz":720},"subject":"Trove doc and CI","message":"Trove doc and CI\n\nChange-Id: I49bd6c824b837d22d9f3945a3e8c2343c831b352\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e2b26fdd7d60b82938682f9c3f82163da9b629cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e2b26fdd7d60b82938682f9c3f82163da9b629cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
