)]}'
{"id":"openstack%2Ftrove~731729","triplet_id":"openstack%2Ftrove~master~Ic9dc4630dd96227eb8a655c3aac223316d5a060e","project":"openstack/trove","branch":"master","topic":"config-sample","hashtags":[],"change_id":"Ic9dc4630dd96227eb8a655c3aac223316d5a060e","subject":"Support to generate config sample file","status":"MERGED","created":"2020-05-29 11:14:43.000000000","updated":"2020-05-29 21:57:41.000000000","submitted":"2020-05-29 21:56:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731729-1590789364572-2d12829e","meta_rev_id":"e15cbb162cf3b2747ce7eb475b4c2e42ac7fb799","_number":731729,"virtual_id_number":731729,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 21:56:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-29 20:58:22.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-29 20:58:22.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 21:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ace88f1fa9eab6a7a823e9bab082447e4cb479ec","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-29 11:14:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a90eda8dc9ffcb69d8a04ad1e123f3e75d07857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 12:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/000d05d117f04e278d7d85349e0ce8d5 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d9c6e1d2d9a4168867accb00f3080a1 : FAILURE in 5m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54293677b3ae4a60a197f5822a66b899 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46eaac519571445aadc072eb31a6e106 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0cf25f689f14427a7343374e8b04fac : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b3430187484dad87ddf442fdc23218 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d39dd93753b4b00b577fa578bcee7c3 : SUCCESS in 6m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be7b626dbb0b475ca8f0f2f115330d0a : SUCCESS in 6m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/271b435af79e465f8030364786731096 : SUCCESS in 6m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e1aa58022f0945cdb9567d5fc75726df : SUCCESS in 1h 14m 14s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8caa4e9676d54a4cbcc9f22148101894 : SUCCESS in 1h 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77ec4b3e0ecaf3c8d56a017cfb30d3f2f83a5e3c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-29 20:58:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd62e71dd590b7f1d86ecbb4e3b9b1d40144ad89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 20:58:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a8c530949d6f3aa22a05b7376a8a1f8a731a5f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:56:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2c0c1473603840bf8bd6f141a4478e2d : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b81aa0b4dfe34537bf63ab50e439527d : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2141a7fe5d41bb9668940b3984ea7b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2c122a0a04649b1b7f253fb19e2d8c5 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11d9eba26b224142b342c3ca1e9047cb : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2e5ddf71294a0c9fc1da4e280e387e : SUCCESS in 6m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d5e742e6fe514fb8bc57f6dad9defeb2 : SUCCESS in 8m 28s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/824a9d9a37374e87b29af782b8c67c6c : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d81644055f6a31d969c1f614ec3ad2f812f20490","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:56:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5f726b73f6ae612e953448ddcd84a91a27bf5a2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:57:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cb41d07851442ab1b5ec7d95ab845b : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87b65e3c0512916ef594c7eb8dda614eb4cfc952","revisions":{"87b65e3c0512916ef594c7eb8dda614eb4cfc952":{"kind":"REWORK","_number":1,"created":"2020-05-29 11:14:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/29/731729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/731729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/731729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/731729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/731729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/731729/1"}}},"commit":{"parents":[{"commit":"5011892b37161e4e84b25a1ec0b808ebaf40fa16","subject":"Doc: update trove design","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5011892b37161e4e84b25a1ec0b808ebaf40fa16"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-29 11:00:27.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-29 11:13:29.000000000","tz":720},"subject":"Support to generate config sample file","message":"Support to generate config sample file\n\nChange-Id: Ic9dc4630dd96227eb8a655c3aac223316d5a060e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87b65e3c0512916ef594c7eb8dda614eb4cfc952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87b65e3c0512916ef594c7eb8dda614eb4cfc952"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
