)]}'
{"id":"openstack%2Ftrove~731734","triplet_id":"openstack%2Ftrove~master~Ia8b5828dfabd595d0ac33a2fdf1d9959f2d1d9ba","project":"openstack/trove","branch":"master","topic":"remove-ds-params","hashtags":[],"change_id":"Ia8b5828dfabd595d0ac33a2fdf1d9959f2d1d9ba","subject":"Support to remove datastore config parameters","status":"MERGED","created":"2020-05-29 11:46:55.000000000","updated":"2020-05-29 21:42:32.000000000","submitted":"2020-05-29 21:40:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731734-1590788453028-002362fb","meta_rev_id":"64528f12ed9008de8bde3da13341f9f7d43e320f","_number":731734,"virtual_id_number":731734,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 21:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-29 20:57:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-29 20:57:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 21:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71cc01b0217de42f2cd3c48df6a4c474e7089680","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-29 11:46:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008ce721870e22ab1bc7e689f9f2d67bc5802073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 12:54:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e6184b767e430bba1f3800d0846552 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7342f326f65b435985a731a4c3829018 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e425868e4fe4bcf8b603fa3f5cdceb9 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ce2efa6001f4bc795f5571af921c120 : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a312650a7cdb4e7bb18d2f0f074d6068 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ea0815ec064122be79c35c469159d5 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bf517372d9b4aa5b8e5a9b785cf2a1e : SUCCESS in 6m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfcac84b9c6b4a6da1f7b06b8124c0b3 : SUCCESS in 7m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/083675fd719f494f91ff69cb064a3880 : SUCCESS in 6m 56s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/36ea0d0a4a4c473caf92dbd14711f42f : SUCCESS in 57m 31s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f836064cfa0f49a996e239ec11a70c96 : SUCCESS in 1h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd687daddf7812740d39f4cdcf5818c13293e5bd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-29 20:57:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d91d86d51ffaf568184e986986448e397a3c6939","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 20:57:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e85efa458f5b471a0b11d076d9c3ed1af5f45ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:40:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c8001dcaddf4ff7b97ea2df80c56be6 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da0741fb6595498a878a2b85f58e678e : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09ee7b4ad8704eaf8a2d197804213473 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e425e968b55f4186ae9fbdee91b3eb1e : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f1b93a091042a79a5e15b28e80b372 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12d5bebc2fea436d9422380ef81c5368 : SUCCESS in 4m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2822f80592145bc8e0a0b36418012c1 : SUCCESS in 7m 59s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5f60166274544b07bcc9451fca58687b : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5f474db697b7168e1afa5aa1a4ac0c72d2664f60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:40:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58a642ecd2eefd96fc8ce76fa791e714e94d93c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 21:42:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59d2e685a8940da96d9d0b53aee5a16 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e76a07ee3c4990bb0b8a156b1c1970 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7af0e4df95066c07f4743eee2ec9c43f301c5ae","revisions":{"a7af0e4df95066c07f4743eee2ec9c43f301c5ae":{"kind":"REWORK","_number":1,"created":"2020-05-29 11:46:55.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/34/731734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/731734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/731734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/731734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/731734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/731734/1"}}},"commit":{"parents":[{"commit":"0421c1e4069e78c58a74d94a29c0118d665bf9d3","subject":"Merge \"Trove doc and CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0421c1e4069e78c58a74d94a29c0118d665bf9d3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-29 11:31:11.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-29 11:46:44.000000000","tz":720},"subject":"Support to remove datastore config parameters","message":"Support to remove datastore config parameters\n\nCLI:\ntrove-manage db_remove_datastore_config_parameters mysql 5.7.29\n\nChange-Id: Ia8b5828dfabd595d0ac33a2fdf1d9959f2d1d9ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7af0e4df95066c07f4743eee2ec9c43f301c5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7af0e4df95066c07f4743eee2ec9c43f301c5ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
