)]}'
{"id":"openstack%2Ftrove~732519","triplet_id":"openstack%2Ftrove~master~Ib8f5062094c0ec7766d4c6c6e7c3c8168e15ebd6","project":"openstack/trove","branch":"master","topic":"instance-status","hashtags":[],"change_id":"Ib8f5062094c0ec7766d4c6c6e7c3c8168e15ebd6","subject":"Set status to ERROR if heartbeat expires","status":"MERGED","created":"2020-06-02 10:38:12.000000000","updated":"2020-06-04 02:59:28.000000000","submitted":"2020-06-04 02:57:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732519-1591239443729-0a49205c","meta_rev_id":"8f3ca7afb6b9cba80a4a6c112da4fe176d6f9274","_number":732519,"virtual_id_number":732519,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 02:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-06-04 01:56:19.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-06-04 01:56:19.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 02:57:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"848a3962372e80e37365ca19eb832323b017f7d9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-02 10:38:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9771a07dd047b66abea66807bc2271526ecaf22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 12:01:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7c4f959a4834564a1017cdb4625f52a : FAILURE in 7m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22ec48d6e83c438c9420eb23066092fb : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79357a6d2ab24075883bd26c04a5de7a : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d771f79ea524220b11cf2510bc836be : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b892cf02334343aafaea527cfbfb6f : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baf71c04284b4fb7a2dc849531a51ae2 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6a4e0f539b44de4aeb3ed26eff8b612 : SUCCESS in 8m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33bbb6ebfb1c4d9a83ccec2a23dc1e17 : FAILURE in 13m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6efc45fb2d3a46e0b6581d2e79dd1499 : SUCCESS in 11m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6f20d9f21d7847fb9d4d6efde2ae71cc : FAILURE in 1h 21m 14s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2325f25cbb69405d9c3cc338448839f0 : FAILURE in 1h 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b55261903d109e20d2f79e01451aaec9f55b85e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-02 12:14:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad557f2f4422fe283c03e36eaa9e651c7e12faf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 13:30:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ad75a55e6114586bd10faf3178875f1 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a79bb3cd90a4934aa4679e04edbf99f : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116c3aa718df45c6946c117cf400be81 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d68a407900674749b929cf8e1eaabbf4 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de69e58077874c0a843d5e4e9ff1112c : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f27e834d41043b7b2b51c35aad8de93 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dedf117a7414bd0a97a2282df683112 : SUCCESS in 5m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b23f20e2f030475ebd47d1c59c19fc22 : SUCCESS in 7m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a092ec15af14105b8a6e64c17645ff6 : SUCCESS in 5m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f120bd6ec8734218ba690bad59c477ff : FAILURE in 1h 07m 44s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbde7d53d2d548f1af13214beb537d00 : FAILURE in 1h 15m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"182f2dd6eba20e6b2aeca36c8682969ca184efb2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-03 12:45:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1417f77ce7454d298ca06e3e84b9e59ea251e112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:01:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/794ce4f4e8464bcaae98c3a685812f3d : FAILURE in 10m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/320e1351078b4dd296f7ceedf5d806fc : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ed1dbeecfd4365a5819566b53b3141 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/813fcee785fc4419a3cba57c8c42846b : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db26b89ec0e4438799865b9718505d19 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aedbbc2254764a35b03d9678bc07004b : FAILURE in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/011a9ad080c94bff859a12d65d41290e : SUCCESS in 6m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20b5587220724e3fb7b7fd7c87dcf2d2 : SUCCESS in 8m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4ca0c88953a4f72b81744378abd6d62 : SUCCESS in 7m 30s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6f7894ec06dc4f95ad59ec22ebfe2905 : FAILURE in 1h 14m 06s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c01fc1e0ff8f4c3dab2e4acb56fee880 : FAILURE in 1h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e91bf0164aab5c3f8ed89a17fd5bd03acb169524","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-03 22:12:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d522edb1edb6adefed89e8d07249f0febf8b479a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 23:26:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/803c9e9ddc174d00835b546ba83d037f : FAILURE in 5m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5393d7b02a834c8da9c3556d584ad2de : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dac41d6563504709828729db58883b0f : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8308664abd7549238840e0bc146726b8 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e201486cd318442aabe1b74b27de6599 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa9645a1f49448491839c784e1df721 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/312d5b442a9046f1a555587d06e04fa7 : SUCCESS in 5m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d1ca9a8bfb64b65be834c1516136452 : SUCCESS in 8m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4bb55c5be0b7401ab7b3f053fd2e5cbd : SUCCESS in 6m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f29c8f9231c446a5a37d222e26134abd : FAILURE in 1h 13m 16s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca8a9f6cec844ff18221d5fd7f9bd5aa : SUCCESS in 1h 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9451950b77c370944ba2172ba72d83c4ee723380","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-04 01:56:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7fa4a3e0b14e2a17feab2b3bfcef706ec9f3ab26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 01:56:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dea7c0ec6aa9b7f48cc948e46003d75b56ea6175","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 02:57:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"86f7db424e8c981f1b07504b234bf8e3c80a01bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 02:57:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27e9087a0bd44e63b1c9eea5a10f89eb : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b123c2d14647b29761e983ba535cae : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5315857428a34e7099b71207b858e0d3 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c206def843e4d6a92eb8e8b6724a668 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeb5618be6644cd38aa179bd67a88719 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/961614b3dfe542c0a18671617ae2bc42 : SUCCESS in 5m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/102d9f37c994486290364b959c8d4504 : SUCCESS in 7m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0cd0ad61edaa4f4f94e70f41c4f68127 : SUCCESS in 1h 00m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd9b9091f6a28a7091be1019b4394eb8373079d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 02:59:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7840ab08d52949a5b9550b0a3c72bb42 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30718226ad8244788523b1f11d103790 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dc117d8dd829c64971baefdbaca835aa5c7fb7b7","revisions":{"dc33ef23a23dd44077b4cc72c4906110045c901e":{"kind":"REWORK","_number":1,"created":"2020-06-02 10:38:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/19/732519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/732519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/732519/1"}}},"commit":{"parents":[{"commit":"ff4b6a13397b6b7102cddff70098b23a2d6ab606","subject":"Merge \"Add doc for how to run functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff4b6a13397b6b7102cddff70098b23a2d6ab606"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-30 12:01:05.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-02 09:51:56.000000000","tz":720},"subject":"Set status to ERROR if heartbeat expires","message":"Set status to ERROR if heartbeat expires\n\nChange-Id: Ib8f5062094c0ec7766d4c6c6e7c3c8168e15ebd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc33ef23a23dd44077b4cc72c4906110045c901e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc33ef23a23dd44077b4cc72c4906110045c901e"}]},"branch":"refs/heads/master"},"c473b573e23fa5033fc86b4b977caa9e7da99550":{"kind":"REWORK","_number":2,"created":"2020-06-02 12:14:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/19/732519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/732519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/732519/2"}}},"commit":{"parents":[{"commit":"ff4b6a13397b6b7102cddff70098b23a2d6ab606","subject":"Merge \"Add doc for how to run functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff4b6a13397b6b7102cddff70098b23a2d6ab606"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-30 12:01:05.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-02 12:14:21.000000000","tz":720},"subject":"Set status to ERROR if heartbeat expires","message":"Set status to ERROR if heartbeat expires\n\nChange-Id: Ib8f5062094c0ec7766d4c6c6e7c3c8168e15ebd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c473b573e23fa5033fc86b4b977caa9e7da99550"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c473b573e23fa5033fc86b4b977caa9e7da99550"}]},"branch":"refs/heads/master"},"7e96fadc2ebe1316ff04cd43e81766588a626d31":{"kind":"REWORK","_number":3,"created":"2020-06-03 12:45:42.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/19/732519/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/732519/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/732519/3"}}},"commit":{"parents":[{"commit":"ff4b6a13397b6b7102cddff70098b23a2d6ab606","subject":"Merge \"Add doc for how to run functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff4b6a13397b6b7102cddff70098b23a2d6ab606"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-30 12:01:05.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-03 12:42:09.000000000","tz":720},"subject":"Set status to ERROR if heartbeat expires","message":"Set status to ERROR if heartbeat expires\n\nChange-Id: Ib8f5062094c0ec7766d4c6c6e7c3c8168e15ebd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7e96fadc2ebe1316ff04cd43e81766588a626d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7e96fadc2ebe1316ff04cd43e81766588a626d31"}]},"branch":"refs/heads/master"},"dc117d8dd829c64971baefdbaca835aa5c7fb7b7":{"kind":"REWORK","_number":4,"created":"2020-06-03 22:12:19.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/19/732519/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/732519/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/732519/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/732519/4"}}},"commit":{"parents":[{"commit":"ff4b6a13397b6b7102cddff70098b23a2d6ab606","subject":"Merge \"Add doc for how to run functional test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ff4b6a13397b6b7102cddff70098b23a2d6ab606"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-05-30 12:01:05.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-03 22:12:04.000000000","tz":720},"subject":"Set status to ERROR if heartbeat expires","message":"Set status to ERROR if heartbeat expires\n\nChange-Id: Ib8f5062094c0ec7766d4c6c6e7c3c8168e15ebd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc117d8dd829c64971baefdbaca835aa5c7fb7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc117d8dd829c64971baefdbaca835aa5c7fb7b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
