)]}'
{"id":"openstack%2Ftrove~732526","triplet_id":"openstack%2Ftrove~stable%2Ftrain~I01c11ccb247a018ae85d7f369f80c38f7e9b7b7c","project":"openstack/trove","branch":"stable/train","hashtags":[],"change_id":"I01c11ccb247a018ae85d7f369f80c38f7e9b7b7c","subject":"Fix CI","status":"MERGED","created":"2020-06-02 11:16:16.000000000","updated":"2020-06-03 00:47:06.000000000","submitted":"2020-06-03 00:45:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732526-1591145153796-6506d538","meta_rev_id":"e8ff90e78abf714205fd7d0fd6712e08a1b1581e","_number":732526,"virtual_id_number":732526,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 00:45:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-06-02 22:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-06-02 22:38:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 00:45:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2325194f33f069e4a57217671847c17c834a8439","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-02 11:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb9f254a2fce11057fa37670388ead2f5e3137c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 13:54:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd67bacce7ff426a8b3f462ee64033b1 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd85e2e42c50451f88cf9ff90f57021c : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac23ad06e6ac4c958f883967c174cd91 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50e16de9377a46ada4b3ed486a217a3f : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc337af4919641448ef8a38aa47d8797 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eaf80ba2157c4d9d8dcc8c5da0344ad3 : SUCCESS in 11m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9236d4c8960e4131a8382a0f70dd587e : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa0ea90f0a1b476fa6c3b4658bdb0480 : SUCCESS in 9m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3c6541ab25f94bbdad37ad21ce56d4ad : SUCCESS in 6m 15s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/85a1a2e1006849d28d9e7ca32ff4f041 : SUCCESS in 2h 31m 33s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b1f51d023b954ef9a9745e56fa83d19f : SUCCESS in 1h 51m 59s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a8473ed511bc49ea92b6e51a98373c82 : SUCCESS in 1h 05m 45s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/f4ff962181094cb1b4093cb59df8a858 : SUCCESS in 1h 37m 14s\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d3521b49cec2416daffc14c9243d8aa5 : FAILURE in 1h 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad18b783ca3d0eb0a4a669a429472cddc9ff26ce","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-02 22:38:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0cb2f26bdeae7bcb31c7ad9a323ec3656758951","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:38:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0058bc29d115249d3834b2a4e9afe3f234f3c448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 00:45:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64ee24b4520f47a5aaa040b44a066e20 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dc601a5ff154395943128dd15408861 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57f61dae1a924e089a8dfec8bdfc574c : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255d52c65f6744acb69f3076cf2979fc : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d28e05eb5dc244d494cefadb8a9881a4 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3712becbe16d4b458b7f005ee4d5fd80 : SUCCESS in 5m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/48c2f9a28bc6491c816b1aca57e06aa7 : SUCCESS in 8m 26s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0e74d61356134ef0815a8bf78b208576 : SUCCESS in 2h 06m 23s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/a3a087bd1b72443a97936340a52983aa : SUCCESS in 1h 34m 29s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/239fdf3709e745df86d3ee9d474a840b : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c143fece2e6d5ba4e214fa4fdc02fc405a958ea1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 00:45:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8ff90e78abf714205fd7d0fd6712e08a1b1581e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 00:47:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dcdd0ef3b6345a7a41447e595bd7d8c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8b5525fd2a544c264a86edcd0d3493709f262af","revisions":{"d8b5525fd2a544c264a86edcd0d3493709f262af":{"kind":"REWORK","_number":1,"created":"2020-06-02 11:16:16.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/26/732526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/732526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/732526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/732526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/732526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/732526/1"}}},"commit":{"parents":[{"commit":"689f33b0ba2c3b18106d588b7ce162730fd7e0c3","subject":"Merge \"Fixes \"a2ensite\" command arg and adds mod_wsgi package installation\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/689f33b0ba2c3b18106d588b7ce162730fd7e0c3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-02 11:16:01.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-02 11:16:01.000000000","tz":720},"subject":"Fix CI","message":"Fix CI\n\nChange-Id: I01c11ccb247a018ae85d7f369f80c38f7e9b7b7c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8b5525fd2a544c264a86edcd0d3493709f262af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8b5525fd2a544c264a86edcd0d3493709f262af"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
