)]}'
{"id":"openstack%2Ftrove~733759","triplet_id":"openstack%2Ftrove~stable%2Fussuri~I2fbc295151a916410ae5b3adb18ca8209a530a22","project":"openstack/trove","branch":"stable/ussuri","hashtags":[],"change_id":"I2fbc295151a916410ae5b3adb18ca8209a530a22","subject":"Raise error when returncode !\u003d 0 and have error","status":"ABANDONED","created":"2020-06-05 07:32:48.000000000","updated":"2021-11-12 10:47:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"128d03fc1d7ad2f9c303f5678677c0f1a057f40c","_number":733759,"virtual_id_number":733759,"owner":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-23 05:56:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-22 10:26:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-23 01:05:08.000000000","_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 10:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 05:56:37.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"cfb1cf7487748499450419e8bc57f9aaf08641e6","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-05 07:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dffd7943c2a78824ec90c988bbb61a0834762f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 10:04:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/689325997a324bc68e02a999b0577b58 : FAILURE in 6m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19d447d354714ea9a8cab334e120e648 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/884da833cc2e4ec8847e854902a772d4 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/126d2b448e4744fdbbe2dd88d1493cde : SUCCESS in 7m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0687a9d59cb9489f8bd38d7bed1fa4f8 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be43a69a06d849bc84e2bf6b44e9ef15 : SUCCESS in 6m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/00caa1c799ab4e568e7270a8f6370361 : SUCCESS in 9m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/699e4b6f7d6d4ee78bde9bd4be812fba : SUCCESS in 6m 35s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/51cc91e0ffea4d958d794d2276daf945 : FAILURE in 59m 48s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/cb184938ae934acbb797221d9023374b : SUCCESS in 1h 54m 08s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/8724903f94f54eda84fda9b509874eca : SUCCESS in 2h 26m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/84ad5ad0080442b58362691491f6265b : SUCCESS in 1h 32m 51s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fdd94d38f4634e63a1832296fa9fc8d2 : SUCCESS in 2h 14m 35s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/02f7265147114c579be90b66d309eef7 : FAILURE in 47m 23s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f1796935ede947daad9c10f69fa21a08 : FAILURE in 50m 48s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/102c45ebc4a54ef996bd43a6f95dae85 : FAILURE in 57m 51s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a201c97f09d452bbccb9641a0ea5506 : FAILURE in 54m 13s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/5a8374178686457c92fcfbb25cc1a84e : SUCCESS in 2h 05m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bec0f25964b8ac7b9ae2aa95b6b10fb86e448ce1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-06 04:40:01.000000000","message":"Patch Set 1: Code-Review-1\n\nthe story number is wrong.","accounts_in_message":[],"_revision_number":1},{"id":"0b7311d6ff6c6f4a0d99ea8fa6fd9d0272911fba","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-06 08:19:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e87958226652e4c03c078cbc73eca7acc2d673ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 10:51:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/525203ed31744dac81bcfd78c420b1f5 : FAILURE in 6m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4bb136bb761446f8767160401184184 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760e595e25be4acf9e2f5a6352f1f74d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4018e131c3d6416eb90b37867da613dd : SUCCESS in 10m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77315482f4364a98a4abaf9c70059e39 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad0a698e4ef4efbb41ee821133208b4 : SUCCESS in 5m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bff2bdc30c1441e3a9e6948372048ea7 : SUCCESS in 8m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/39cd75e3bb424deea6bf19325f3aa6a0 : SUCCESS in 7m 46s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c798acab237545c3bd724673cfff4d6d : FAILURE in 57m 07s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/9acead7b712444068bb598cedf9f7e9a : SUCCESS in 2h 13m 19s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/32b34ce505d24c83ae67a34bf45a1696 : SUCCESS in 2h 27m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/30f0b535c7c4448d8489737f78393e16 : FAILURE in 1h 18m 34s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/892ef9ddbfe4420da4db0436fa9bd99f : SUCCESS in 2h 05m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/b453c7cbf44647de942659f5d23e54c1 : FAILURE in 53m 27s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/0ef84413e1d6402bad90e485dd17ff1c : FAILURE in 43m 58s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/dc76245060e44aa6b86d44e53a1b05f9 : FAILURE in 1h 16m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b980c3b932514effab5381856fc04def : FAILURE in 59m 37s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/37c3e80431b94ecf8a0c3b7e7e48a7c4 : FAILURE in 56m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"261c275ebb8af9f7664c7cc6b76d0fb77e62352a","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-06 10:53:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"85b89f0973097d27de9d629fc001fd2c549f8e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-06 13:37:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/512aabfd343d437a8077683ddf2fbdbc : FAILURE in 5m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69462d4e4a964630ba1e245acdcfd2b0 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae0ab7b87414131b3b0c2ec07561e6a : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95f81994a0f14ce981dd59ea3572ef0a : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6865b638b7fd4eeb817c6dee732115bf : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fadce54d6bc4b418e27bad4a94b9a41 : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e37e49a9262b41859d3531a092b906e8 : SUCCESS in 11m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5890660688184ae4afcd2f2522d63656 : SUCCESS in 9m 02s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/21c409a60029497596985e7516bd476e : FAILURE in 42m 47s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/405301daec6b49e7a04e95ce2d0fc518 : SUCCESS in 2h 41m 47s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/547b29599cac48cba7c11d78f4bdcfc8 : FAILURE in 38m 39s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/454a6f527d9144b4a826cdcf300f82bd : FAILURE in 1h 37m 41s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/8fc999f0816346bb9060a5aee29f8043 : FAILURE in 1h 42m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/de5783bf66dd4e47b91cf56e1cbcf1f9 : FAILURE in 51m 56s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/d2c48109b4644010b4b8f41c4bcc3653 : FAILURE in 41m 59s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5cb5e46c2e2a49c885b98688432b8864 : FAILURE in 40m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b8e3f6fa74645e1a0e8b6ce30ac9870 : FAILURE in 1h 22m 06s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/b46aa52897ad4a85823e2b9d3f7c16c7 : SUCCESS in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7d2a0ddd662bf84e3f64bbf0c59a02874cd4e6cb","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-08 01:08:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c49eb158ace8860c86aa67f3b238599ad9721fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 03:48:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/070bb2cca8104ed4b27b734bf8cda0d0 : FAILURE in 7m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a033bef6c9f4283a6212013a7a64eb0 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19b4761a7674b859c966632ae0b2a1c : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/139091980401418281603cc294b69489 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf99e776f2c647abaf13fbca7338b16c : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f246fae51c4b7d9d2699f619c4aee7 : SUCCESS in 5m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/004915160f8c4489b2cef2f31a23c1b6 : SUCCESS in 8m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c4a41cf20d3f4c8e810a56bce10de723 : SUCCESS in 8m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/113aae46283745eeaf901816e8e79ec8 : FAILURE in 1h 03m 18s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4f49668f2e304775a66d0a98b67485b3 : SUCCESS in 1h 59m 31s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/02be54e9a66340f094862b9caed66289 : FAILURE in 2h 38m 06s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/ff30c6e5dd6b40279bcddb802d216003 : SUCCESS in 1h 34m 25s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/fb6a9880eb4f419a87b92c051de9b5ee : SUCCESS in 2h 10m 27s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/7c494d215f4c4cd79c056b107827aabf : FAILURE in 1h 05m 11s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4f6d19ebd8f041f4bc3ecf5eba9e5e57 : FAILURE in 43m 13s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/888cdba65a7f4e679b0ae48af2f654a5 : FAILURE in 1h 05m 27s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23795d0237a94c388908fd635bf4e5ad : FAILURE in 57m 28s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/79fbbc323eb240ed83ab9e9349f76d39 : SUCCESS in 1h 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f32c42bed9999814a244d474b73f26a6632f4247","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-08 03:49:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5167c3f0c72fa8cc2a4a6a144d2d569fb9d7ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 06:09:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5895df1425284692956f8d0d49198d1e : FAILURE in 5m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba85dd25829e4218b97768a67ece72b8 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc048c8ac1644d4bad59db897742e085 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a10ddaeb582402399ed445bb2c92b7f : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61911e3b9f6745baa648851ec0dd931b : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7229ce524b6a4121a50c87d9ca703d6b : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cee31515ccb42669205cb664b004308 : SUCCESS in 8m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b44d5635b77475a873484a3b05913c7 : SUCCESS in 6m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/989a15d02c4a41ccafa677dc8349e648 : FAILURE in 54m 15s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/47697f2c302e470d924d173d77da5a25 : SUCCESS in 2h 06m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ba5382ce995542be9f596d9a27367120 : SUCCESS in 2h 14m 48s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/a1328b433caa41dbbac9e07493180a49 : SUCCESS in 1h 24m 55s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/b4279e5b0d8c472f88fe9305e4035d6f : SUCCESS in 2h 06m 56s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/2de67ee685bf4c91980ed2093a1794ff : FAILURE in 42m 59s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/8c5dec36f1c84ff4a39febc9b0e9feb3 : FAILURE in 40m 03s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/984336f87d9947fb979bf636781276a9 : FAILURE in 1h 09m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2f02ba14cc8479f8f9e31cb475f6923 : FAILURE in 47m 51s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/7c4c7460d9fe428caa450c4711ec3a4d : SUCCESS in 2h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15de632c96a006d0f49bcd171b52b22032f3d2d5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-23 10:10:58.000000000","message":"Patch Set 2:\n\nCould you please see my questions in https://storyboard.openstack.org/#!/story/2007758?","accounts_in_message":[],"_revision_number":2},{"id":"017a9896492cebf31647f0e2b8833c52fd27e2f9","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-10-22 08:02:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd87a1b52290b152aa33d6c44f28a8f43c791059","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 09:27:38.000000000","message":"Patch Set 3:\n\n@Nguyen, could you please talk with @Sam Morrison in https://storyboard.openstack.org/#!/story/2007758 for a fix that could solve both your issues?","accounts_in_message":[],"_revision_number":3},{"id":"2861eb608214a149b53065c9eb8092ef9f9d60ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:26:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7569b4967cc4d85806d156e39a49439 : FAILURE in 7m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c85cf35a33f475eb897ff1f75bd6e61 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097351d92d79460e80848a9664c87bdb : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41feceecf03a49589525c661b1c411c2 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da3146a328c04d839affd2c189b66ff9 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929180aade494333856744617e4ee310 : SUCCESS in 7m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a0804a5f2dde45c8914cbd3f39d786cd : SUCCESS in 7m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f826a16201dc4a53bf9b84d486158c94 : SUCCESS in 53m 18s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/14adad45786f475f9afded6fb52728dc : FAILURE in 7m 00s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b106d250f916429e943288c4d90c6727 : SUCCESS in 2h 18m 05s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d65a233ad0ba4705a3b6823e1f428892 : SUCCESS in 1h 32m 44s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/76dee80dd25549f88582a3cc4e12010e : FAILURE in 5m 58s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/659053d03f4e44978c0a9bd780d5b32e : FAILURE in 48m 23s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/fb13da12216c488c99dce070fa2e70bf : FAILURE in 44m 30s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/a85fb91ff0864a478d373ffce75e1ddc : FAILURE in 1h 00m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01bac12e55b2411a8bc1bf3560d3c39f : FAILURE in 46m 15s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/9ed163627ae746e6ba2d04358616f883 : SUCCESS in 1h 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e9b5494269ed83be1a76360d0a34de6c6980e2d0","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-10-23 01:05:08.000000000","message":"Patch Set 3:\n\nI think we are have the same error when backup \u0026 restore in ubuntu bionic.","accounts_in_message":[],"_revision_number":3},{"id":"f76abeb79d7377208a17b65f304c515b91e7c49c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-23 05:56:37.000000000","message":"Patch Set 3:\n\n\u003e I think we are have the same error when backup \u0026 restore in ubuntu\n \u003e bionic.\n\nI know, you both submitted patches.","accounts_in_message":[],"_revision_number":3},{"id":"2e1c80a3b209a1fbfeeed04029fc0fe04a310c28","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2021-11-12 09:49:20.000000000","message":"Abandoned\n\nnot effect from victoria","accounts_in_message":[],"_revision_number":3},{"id":"128d03fc1d7ad2f9c303f5678677c0f1a057f40c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-11-12 10:47:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2e8021a4de1a7876dd40bbe8ba37dac0959cb560","revisions":{"3b8b8f20597074406ede60531570b8feaf30b16d":{"kind":"REWORK","_number":1,"created":"2020-06-05 07:32:48.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/59/733759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/733759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/733759/1"}}},"commit":{"parents":[{"commit":"8d9b938a5596db02d42090104f58a53d6aa4bdde","subject":"Fix CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d9b938a5596db02d42090104f58a53d6aa4bdde"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-05 07:07:48.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-05 07:32:45.000000000","tz":420},"subject":"Raise error when returncode !\u003d 0 and have error","message":"Raise error when returncode !\u003d 0 and have error\n\nChange-Id: I2fbc295151a916410ae5b3adb18ca8209a530a22\nStory: 2007587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b8b8f20597074406ede60531570b8feaf30b16d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3b8b8f20597074406ede60531570b8feaf30b16d"}]},"branch":"refs/heads/stable/ussuri"},"2b498788481efa3cf21e5561eea217f2b5ce32de":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-06 08:19:15.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/59/733759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/733759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/733759/2"}}},"commit":{"parents":[{"commit":"8d9b938a5596db02d42090104f58a53d6aa4bdde","subject":"Fix CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d9b938a5596db02d42090104f58a53d6aa4bdde"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-06 08:14:36.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-06 08:18:23.000000000","tz":420},"subject":"Raise error when returncode !\u003d 0 and have error","message":"Raise error when returncode !\u003d 0 and have error\n\nChange-Id: I2fbc295151a916410ae5b3adb18ca8209a530a22\nStory: #2007758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b498788481efa3cf21e5561eea217f2b5ce32de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b498788481efa3cf21e5561eea217f2b5ce32de"}]},"branch":"refs/heads/stable/ussuri"},"2e8021a4de1a7876dd40bbe8ba37dac0959cb560":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-10-22 08:02:10.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/59/733759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/733759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/733759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/733759/3"}}},"commit":{"parents":[{"commit":"a5a5c5e060c37c4d0ec5fff125253ca02c1f1817","subject":"Handle case where no networks are selected at create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a5a5c5e060c37c4d0ec5fff125253ca02c1f1817"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-10-22 08:00:55.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-10-22 08:02:07.000000000","tz":420},"subject":"Raise error when returncode !\u003d 0 and have error","message":"Raise error when returncode !\u003d 0 and have error\n\nChange-Id: I2fbc295151a916410ae5b3adb18ca8209a530a22\nStory: #2007758\nTask: #39966\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e8021a4de1a7876dd40bbe8ba37dac0959cb560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2e8021a4de1a7876dd40bbe8ba37dac0959cb560"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
