)]}'
{"id":"openstack%2Ftrove~734794","triplet_id":"openstack%2Ftrove~master~Icc4eece2f265cb5a5c48c4f1024a9189d11b4687","project":"openstack/trove","branch":"master","topic":"user-ip","hashtags":[],"change_id":"Icc4eece2f265cb5a5c48c4f1024a9189d11b4687","subject":"Support subnet and IP for instance creation","status":"MERGED","created":"2020-06-10 10:51:23.000000000","updated":"2020-06-10 23:06:57.000000000","submitted":"2020-06-10 23:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734794-1591830333264-1e1c92d6","meta_rev_id":"2c391e75a1bbb6508e6b21cdd200cabc4a723668","_number":734794,"virtual_id_number":734794,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 23:05:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-06-10 21:22:28.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-06-10 21:22:28.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 23:05:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe970f476497900a52145b9a08198e9dca0be4ac","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-10 10:51:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dfc12994ba96667be34f2ca039604934274696b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 12:04:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/baaa4513fd574750895202c4f0da7e77 : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/488543cdd3984f998e35364d494a44df : FAILURE in 6m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03a5abcaa3b14900996665dad40a7611 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68950a1f95be447ca147c2e732c661c6 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/145abcee65234edf9e6642ef82d4049c : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67e31cbddb20467faa0559769b63e6ff : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab03bffb8f914a74a40e32e64e18a029 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b53e7c73c5ff4352b7686b2b5b3bdf07 : SUCCESS in 4m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2efb36636514460e81a6003148f50139 : SUCCESS in 6m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/000aa49c97a248aa85756d877474f373 : SUCCESS in 6m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/572cc3f68a5b40468a88c9c37f0e855a : FAILURE in 1h 00m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f04691ce7f3d4fbe9d52da4460be8a2c : SUCCESS in 1h 08m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00831dea12332dd61bf3ff05123601a0b0421d41","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-10 21:22:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c225ec67ed70fba211997795fc9e7d83fd25103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 21:22:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0471712d28d5c9b81e857b7997539550de48ae32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:05:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9c767e13e025442ca042c71e2ed4b8ae : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28de86b75d3d4ae48f63ec52b6848125 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60451bb890c64488a884e5d4ba882114 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c611f1cc2f74f9da5a1f91d5d45a32b : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efc2bdddc23c4414b6d751a03d8ab4ef : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/894562760fd7434f87ab804b4008fb13 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44b99da6d9d041cebe9863e784afde0b : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c82ab6c152d641f899fe3ad3c79365e5 : SUCCESS in 9m 43s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d89522110e0542368c747a229a84501b : SUCCESS in 1h 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52ddd802ffd6c5264e1798ef262c6003db904bc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:05:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"757c10900214d36953acc975a5046c66418a9c25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:06:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/eacaf77c54f44da49a23ebc645c86c70 : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b95aeecda147848a2591617166fd16 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efafffa923a54ebab5b938214459f86c : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"535417240709694080255c064c9d84a5ab1d5372","revisions":{"535417240709694080255c064c9d84a5ab1d5372":{"kind":"REWORK","_number":1,"created":"2020-06-10 10:51:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/94/734794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/94/734794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/734794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/734794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/94/734794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/94/734794/1"}}},"commit":{"parents":[{"commit":"b77f7b9fe764f519a7dbfd9026b15db320d4aebd","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b77f7b9fe764f519a7dbfd9026b15db320d4aebd"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-08 23:09:22.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-10 10:41:35.000000000","tz":720},"subject":"Support subnet and IP for instance creation","message":"Support subnet and IP for instance creation\n\nSupport ``subnet_id`` and ``ip_address`` for creating instance. When\ncreating instance, trove will check the network conflicts between user\u0027s\nnetwork and the management network, additionally, the cloud admin is\nable to define other reserved networks by configuring\n``reserved_network_cidrs``.\n\nChange-Id: Icc4eece2f265cb5a5c48c4f1024a9189d11b4687\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
