)]}'
{"id":"openstack%2Ftrove~734796","triplet_id":"openstack%2Ftrove~master~I909cd37b34466f41ae4cca0e85cb8a917ba163cc","project":"openstack/trove","branch":"master","topic":"list-show-replicas","hashtags":[],"change_id":"I909cd37b34466f41ae4cca0e85cb8a917ba163cc","subject":"Show replicas info for listing instances","status":"MERGED","created":"2020-06-10 11:07:07.000000000","updated":"2020-06-10 23:21:46.000000000","submitted":"2020-06-10 23:20:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734796-1591831210863-2acfc7a0","meta_rev_id":"a06c31bb8eab79409bd83511f6f71c1a65f6976d","_number":734796,"virtual_id_number":734796,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 23:20:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-06-10 21:25:50.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-06-10 21:25:50.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 23:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"186a6d45f2e60fe0e59e4089b9fb97e1dbe093b7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-10 11:07:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"816cbc9485f2daef0684e860029701697de47b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 11:51:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd94e9c044494df9b747bcf404469234 : FAILURE in 5m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff8ab826a3c44bf3a2913c0eef33f047 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09832289d9b47059d5002303f4f9978 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ec857d11afe483eb87a8fdf6c68ca56 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d9eb908e9c041838b585a27b81aad33 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df58b9924184b3499d97dfc7e408273 : SUCCESS in 5m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/257a7771ada74c458c95b56c14399383 : SUCCESS in 6m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d62c33349aff4149ba881ac5819694a3 : SUCCESS in 5m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0de794cbec2f4a39bc39b2228cc879cd : SUCCESS in 40m 25s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f07762a988e45178aa3ec9f7b978ea2 : SUCCESS in 42m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"474386634294d4368d0b318e7859dd61f827b80f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-10 21:25:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f71b07be4a7881a6ce6a74d549f8039e98c85e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 21:26:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ea84f18791a68ba489e38c340a0fcf71e646113","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:20:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f87e06040a4541f7a5801e5591bcaadc : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a510ca399374f629b6d68e6dc40d6cc : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4b4cd713ed4f49af8cbdb231ed2e76 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aadd25d9fd034a689cf906f6724d5f8b : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff6e1f9c5194e5294b320a1eac89d86 : SUCCESS in 7m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d667407d07a24f738e22d8050df2d364 : SUCCESS in 8m 18s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f8c64768685440d898bc4b8194a1f608 : FAILURE in 1h 47m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1a0450355b86d82fe17821e3a8ad6835d759d7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:20:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1094ffcf69343bcdfe10df7ee9fd2672d1b98d76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 23:21:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d093fcdfdf64275afc176e7502d70ec : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb70d24c3f2e7edb13f0c687958b4a7e96f4cd39","revisions":{"fb70d24c3f2e7edb13f0c687958b4a7e96f4cd39":{"kind":"REWORK","_number":1,"created":"2020-06-10 11:07:07.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/96/734796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/96/734796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/734796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/734796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/96/734796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/96/734796/1"}}},"commit":{"parents":[{"commit":"535417240709694080255c064c9d84a5ab1d5372","subject":"Support subnet and IP for instance creation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/535417240709694080255c064c9d84a5ab1d5372"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-09 03:50:08.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-10 10:51:37.000000000","tz":720},"subject":"Show replicas info for listing instances","message":"Show replicas info for listing instances\n\nChange-Id: I909cd37b34466f41ae4cca0e85cb8a917ba163cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb70d24c3f2e7edb13f0c687958b4a7e96f4cd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb70d24c3f2e7edb13f0c687958b4a7e96f4cd39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
