)]}'
{"id":"openstack%2Ftrove~735405","triplet_id":"openstack%2Ftrove~master~I29b4deaf0a42d79f12b9ddbb53724a5f4ee3f8ac","project":"openstack/trove","branch":"master","topic":"bug/1882877","hashtags":[],"change_id":"I29b4deaf0a42d79f12b9ddbb53724a5f4ee3f8ac","subject":"Reference Implementation using new oslo.db\u0027s enginefacede","status":"ABANDONED","created":"2020-06-13 00:08:43.000000000","updated":"2026-08-22 04:20:09.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f45a3b3b1c5319727767ca26ea78289341953105","_number":735405,"virtual_id_number":735405,"owner":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-13 00:46:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-13 00:46:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce83f7cd0da67d9e8259af0c00553bf6e0ad764b","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2020-06-13 00:08:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b686b3516304dda783b4a072ca65e256cd1cbea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 00:46:10.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23c9efb1f9fb4766ae7f20f0c9803bae : FAILURE in 6m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4f559e3d3494b88be9e7e30250c6d2c : FAILURE in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af209bcaf2b54fd4b080b3a8432628ac : FAILURE in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a36cb86910154baead04277477dd2688 : FAILURE in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90940a8fb5af4dddb315a6186ecd08b5 : FAILURE in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf45f22289ce4abe805a0c3403b90133 : SUCCESS in 6m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f9edc20d79794907b1bd05010f182865 : SUCCESS in 8m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1bd9fc9741b84a86af448e240d4657af : SUCCESS in 5m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/432279a07563410c90203ad957687233 : FAILURE in 34m 44s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1210436a46a4977aba40639ca964681 : FAILURE in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f45a3b3b1c5319727767ca26ea78289341953105","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"date":"2026-08-22 04:20:09.000000000","message":"Abandoned\n\noslo.db already updated.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52fa0a80e517a4205e0412bd1fbea732a605922e","revisions":{"52fa0a80e517a4205e0412bd1fbea732a605922e":{"kind":"REWORK","_number":1,"created":"2020-06-13 00:08:43.000000000","uploader":{"_account_id":31737,"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","username":"hiwkby"},"ref":"refs/changes/05/735405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/05/735405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/735405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/735405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/05/735405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/05/735405/1"}}},"commit":{"parents":[{"commit":"428e3644f8770a608f6e4dfc617f45f6e17cb685","subject":"Support image type for guest image building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/428e3644f8770a608f6e4dfc617f45f6e17cb685"}]}],"author":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2020-06-12 23:57:51.000000000","tz":0},"committer":{"name":"Hirotaka Wakabayashi","email":"hiwkby@yahoo.com","date":"2020-06-13 00:05:18.000000000","tz":0},"subject":"Reference Implementation using new oslo.db\u0027s enginefacede","message":"Reference Implementation using new oslo.db\u0027s enginefacede\n\nI try to add a single call to enginefacade.get_engine() and check\nif it resolves my problem for some evidence. This code change might\noccur DuplicateOptError and we need ask oslo.db team to help Trove\nor we should take another approach like singleton oslo.config that\noslo.db\u0027s Mr. Hervé Beraud has suggested.\n\nChange-Id: I29b4deaf0a42d79f12b9ddbb53724a5f4ee3f8ac\nRelated-Bug: 1882877\nTask: 40054\nStory: 2007498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52fa0a80e517a4205e0412bd1fbea732a605922e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52fa0a80e517a4205e0412bd1fbea732a605922e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
