)]}'
{"id":"openstack%2Ftrove~735498","triplet_id":"openstack%2Ftrove~master~I953d82c72c0a009d81443433ce301a79150084cf","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I953d82c72c0a009d81443433ce301a79150084cf","subject":"Fix error when create multi replica","status":"ABANDONED","created":"2020-06-15 05:18:04.000000000","updated":"2020-11-30 03:10:57.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f5018fafdf532c8cd6b3a40cd4c7b497f8ad62a","_number":735498,"virtual_id_number":735498,"owner":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-23 12:12:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-07-08 11:53:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":1,"date":"2020-06-25 04:27:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"value":-1,"date":"2020-07-09 23:38:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 12:12:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-25 04:27:32.000000000","updated_by":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"reviewer":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"state":"REVIEWER"},{"updated":"2020-07-08 11:53:57.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-07-09 23:38:52.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"e31a7cca131c57aa00bac0a808887213a403de2a","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-15 05:18:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec47db67c08614e1864078934844eb4b9ccb5662","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 05:35:09.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1211972e344cf0adfa0db421eb1885 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9cb74c5de3d4f6dae3fb7577cd13270 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce052ada0cb3420896d54d5fc299b412 : FAILURE in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8914daa696845528c9fbae50092fbd4 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a72440f66d6c4b49b251d372edc6a47f : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee73ec9cd1ab4f13acfe92b630d27b78 : SUCCESS in 5m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/593a3b7e613447ddb52d112b0e668b91 : FAILURE in 7m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/12e7f71e504c4d1aa015ce8c6b3d7ee9 : SUCCESS in 8m 34s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/11fce0defe3c4c49a9361a8eb8adf35f : FAILURE in 8m 54s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5826e0b8791e4755ab99fc4994b67ec2 : FAILURE in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5eb091d4cf68379702cd3cbee0cb443f42f2370f","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-06-15 06:14:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc8342c757dfa8002ae6dd55108854f98fe4ee34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 06:52:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c5bbf54fb0b4b81b692e264935d0c0a : FAILURE in 9m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/294234d95b5a4a02ba16ce7c0e0f6b64 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab63f4fdc1e142a59fa5a86bd16cd76e : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18fc18ebf3ce4183b363d3e83d94fbf9 : SUCCESS in 11m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5248085a3abe48a386b8cfd7bf5f0df7 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbb48afd1b9451eb1975b8671298a43 : SUCCESS in 8m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/135944d387494972ba2f39f4fc41ed51 : SUCCESS in 14m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/202c03fa54be4f69bc1c3aec07936dc6 : SUCCESS in 8m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c0ca7bbea483445aba19db8ea9289585 : FAILURE in 16m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53f5ac03075749c8aee160b64fbaf93f : FAILURE in 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e37955342d91342e810d9525936e1266577d1782","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-23 10:07:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d1344399f869b2e1f487e994bd4b725cbe99c213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 12:12:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feda97f4b7304ba4b97cc23945dba13a : FAILURE in 8m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf445614fde2461d99581c1e7e7ad725 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd3d97fbd2444258e9d0d6f6f4fb07e : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6990ea8004d410280362bd6f385ce17 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56b03a3ca06f4e9cbdf3bcadfdbdc42e : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ab3b50db504c1bab4da141278b92f3 : SUCCESS in 6m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8b2605c83fcc487fbe21b02297b4bf0d : SUCCESS in 9m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ddf0a60358e14d4fb3cf19e77a0a3437 : SUCCESS in 7m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5e933689f1474e5d9c0ca21fc18286e7 : POST_FAILURE in 1h 58m 46s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73efb179285f44af9ddaf6fbbe3bc4c1 : SUCCESS in 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba9569319ebf5456ecfd47124d9e4f78fb13df17","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2020-06-25 04:27:32.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"134031f4bea82b9fb50f9b6c197944a8a14fbb63","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-29 12:55:50.000000000","message":"Patch Set 3:\n\nPlease see my comment to the story.","accounts_in_message":[],"_revision_number":3},{"id":"67d8795873a5cad56b0216e677edfcb58a0d2a01","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-07-08 11:53:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d624599366419d724223b998dd2a52e16ac1ca48","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-09 23:38:52.000000000","message":"Patch Set 3: Code-Review-1\n\nsee https://storyboard.openstack.org/#!/story/2007812","accounts_in_message":[],"_revision_number":3},{"id":"ccd2ecab736c63dc2fe0382acb0f12b37e218e04","author":{"_account_id":25254,"name":"Nguyen Van Trung","email":"trungnvfet@outlook.com","username":"trungnv"},"date":"2020-07-17 03:20:03.000000000","message":"Patch Set 3:\n\nPerhaps, the bug occurred with mariadb database","accounts_in_message":[],"_revision_number":3},{"id":"2f5018fafdf532c8cd6b3a40cd4c7b497f8ad62a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-11-30 03:10:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62c793c89825d0377ed7947c5bbd3475fd4fd40f","revisions":{"9030726844ed9bc4ba4971144fd130e24bac0754":{"kind":"REWORK","_number":1,"created":"2020-06-15 05:18:04.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/98/735498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/735498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/735498/1"}}},"commit":{"parents":[{"commit":"428e3644f8770a608f6e4dfc617f45f6e17cb685","subject":"Support image type for guest image building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/428e3644f8770a608f6e4dfc617f45f6e17cb685"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-15 05:10:19.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-15 05:17:56.000000000","tz":420},"subject":"Fix error when create multi replica","message":"Fix error when create multi replica\n\nonly create user if not exists or drop user if exists\n\nChange-Id: I953d82c72c0a009d81443433ce301a79150084cf\nStory: #2007812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9030726844ed9bc4ba4971144fd130e24bac0754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9030726844ed9bc4ba4971144fd130e24bac0754"}]},"branch":"refs/heads/master"},"61403b71db2eb6fc86735c5e980d1fb9849b4e81":{"kind":"REWORK","_number":2,"created":"2020-06-15 06:14:58.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/98/735498/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/735498/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/735498/2"}}},"commit":{"parents":[{"commit":"428e3644f8770a608f6e4dfc617f45f6e17cb685","subject":"Support image type for guest image building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/428e3644f8770a608f6e4dfc617f45f6e17cb685"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-15 05:10:19.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-15 06:13:08.000000000","tz":420},"subject":"Fix error when create multi replica","message":"Fix error when create multi replica\n\nonly create user if not exists or drop user if exists\n\nChange-Id: I953d82c72c0a009d81443433ce301a79150084cf\nStory: #2007812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61403b71db2eb6fc86735c5e980d1fb9849b4e81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/61403b71db2eb6fc86735c5e980d1fb9849b4e81"}]},"branch":"refs/heads/master"},"62c793c89825d0377ed7947c5bbd3475fd4fd40f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-23 10:07:37.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/98/735498/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/98/735498/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/98/735498/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/98/735498/3"}}},"commit":{"parents":[{"commit":"d81542481b2c753b0c03664ee7b67cc59356598f","subject":"Doc: Fix configuration API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d81542481b2c753b0c03664ee7b67cc59356598f"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-06-15 05:10:19.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-23 10:07:37.000000000","tz":0},"subject":"Fix error when create multi replica","message":"Fix error when create multi replica\n\nonly create user if not exists or drop user if exists\n\nChange-Id: I953d82c72c0a009d81443433ce301a79150084cf\nStory: #2007812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/62c793c89825d0377ed7947c5bbd3475fd4fd40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/62c793c89825d0377ed7947c5bbd3475fd4fd40f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
