)]}'
{"id":"openstack%2Ftrove~735743","triplet_id":"openstack%2Ftrove~stable%2Fussuri~Ie49b0622d8ae6b9b7189173a02f21a28c8dfca04","project":"openstack/trove","branch":"stable/ussuri","hashtags":[],"change_id":"Ie49b0622d8ae6b9b7189173a02f21a28c8dfca04","subject":"Fix building raw image","status":"MERGED","created":"2020-06-15 22:30:41.000000000","updated":"2020-06-18 07:21:41.000000000","submitted":"2020-06-18 07:18:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735743-1592464687163-50419c5f","meta_rev_id":"8f91637b1c432553d582eb42eb1dc79c0cbaf2d1","_number":735743,"virtual_id_number":735743,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 07:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-06-16 19:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-06-16 19:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 07:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05b74807202db983a3ff15f1aeb7d97ddb5fba74","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-15 22:30:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c3d0f3f7fad71ac81aa32d2fec9059fba178d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 22:50:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9e5e481573c46f0afebfe48c6c47d7c : FAILURE in 10m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ddcf10e69a9443eb1ff8211d9b3d631 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ce965ad8a04327885e260d8dca92f1 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba184b0048e449aa0b6e4ebed8e2342 : SUCCESS in 11m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32969b5fba0844028b163459bfbe6a77 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5782e56be64726a218bd56c9af5ba6 : SUCCESS in 9m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfa8b65aa8a14267991e723b9c14a097 : SUCCESS in 11m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce96b405ff454dd8b2361de59b844bdb : SUCCESS in 9m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3699d4cecd14485cadceecf3b563c9f9 : FAILURE in 14m 55s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/01bc4d67d4ad4cec8801140fa35bdd30 : FAILURE in 13m 41s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/9285655e743841c788f1bce772bc8143 : FAILURE in 14m 52s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d372b1a46ce140b58cae2b12c2d05781 : FAILURE in 14m 33s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/d1da9713d14941dcb07254123a9fb462 : FAILURE in 14m 54s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5f7ba9d4f9d142759c81f53ab9baf6ed : FAILURE in 14m 31s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/f0290164445240e8b583a90a927d50df : FAILURE in 15m 37s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/5345ac4363a341e3b01245fa2303b389 : FAILURE in 15m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efa1107d10554b1f982c07b7074f4e7d : FAILURE in 14m 18s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/a3fc355e61244a9ab06eefe462a77817 : FAILURE in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05bab9cf45a30b8389c7b0dc5da93256a592be0d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-16 19:53:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3f613b1a42382a6e438ba80bc3ce625e1720fcd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 20:22:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ce4e1e3f2c24e399e5987439c9d7630 : FAILURE in 9m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1bfdda7b3164386b80e0911516aed8a : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37746bdc87304da0b3da54fc14f75866 : SUCCESS in 11m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28fa69a927da47a891ceaccc982afa0e : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9b40075c9984ba989b7804642a0a312 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f4a701e88f4697b94f4f53e6274c8a : SUCCESS in 10m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdde0bd88b454a34938601311ad08a12 : SUCCESS in 12m 53s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a657744cc3824279afa2198ddb7bc135 : SUCCESS in 9m 11s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ce91656ab11045f7907e01669aa15d28 : FAILURE in 19m 39s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0252984b16f44b21b484c7f7178b54e1 : FAILURE in 15m 23s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ee3501fb037346dd8b02ad632870686a : FAILURE in 14m 45s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/6b3a719559f84af59435b443f9e70d19 : FAILURE in 17m 16s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/63d505abe98740b6b003421239c6157b : FAILURE in 13m 51s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/11889ec8d14f430a8206a40c4345e4b4 : FAILURE in 16m 14s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/5af927a3242b4fd6842503da5ccd4e82 : FAILURE in 15m 17s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/328848d159ea4d2a8aeffc6e35fd7d12 : FAILURE in 13m 15s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/383fa8967c0948c2bd29cc28b08b4cf4 : FAILURE in 21m 42s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/f9851e986a234683aa7250f864a6895d : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a63d141cab981d9079f663e2a2d77b79213a22a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-17 06:12:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5bdf49faec2e27e481864d001bf54f06462fdf1f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-17 08:20:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3655af235f06ed3ad218618e0f81b26593b47862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 08:34:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63cdfd1d037c4ac4af8b6d09a8c2fd78 : FAILURE in 11m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db12181e940840d58c938e86c6b80946 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c4d61c3f3b74d259bc3b974df10aa8e : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b0a047f7ab487db3467d466b7a4838 : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57973f63fa9345a5b5234617dfbe12c7 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc47ef0b0bbc4438ad1ecdd2c0b6c3b4 : SUCCESS in 10m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81ea9776d7424dfea37797a03d6d83cd : SUCCESS in 9m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f205351fe24d4fe0a9d91a8dde93206e : SUCCESS in 9m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/064a6cf17c5549a9b434b8c96ff2e99f : FAILURE in 58m 33s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/cf2401f5cab64259b6ae8049faae0b37 : FAILURE in 36m 03s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/9c88a542785e4082859709ec66accbc8 : FAILURE in 1h 02m 33s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f33939468cdc4bab85d0ebb18ab73346 : SUCCESS in 1h 42m 53s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/10c178c27dfb4a99bc014d18f8a527f3 : SUCCESS in 2h 12m 49s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e7a01bdc0b414c698260cc57480d676e : FAILURE in 45m 25s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6d543b9ec6904c018baa8fc3b51b5a23 : FAILURE in 45m 22s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/bd50978bae7e44138b39f961e03f3325 : FAILURE in 1h 16m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9272cc8716e94965ac0dfc595e1ecb7b : FAILURE in 1h 02m 12s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/f838bad0eb914882a60837fd20ff1641 : SUCCESS in 1h 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6fd3889638fe1607385368cc81f0fe6c3cea522b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-18 02:49:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9203db3f9f45c99559b9b7a975c52b0bdc9ae02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 05:03:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d97d7040ed0424981ea8c55719d3f9f : FAILURE in 6m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f39287d03c542c38d4f6c179633cbba : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541a6f466a944d78a55253b836d1ad8b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a8bac5dd4bd471caf25f62abcb62187 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dab2812d48ac4f2bb6ae0e2e67c34f70 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798d8d7ff063427385272d4139dc5651 : SUCCESS in 6m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0af6bd57ebfb4ed78b0d8e098d127af5 : SUCCESS in 8m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e66c648acf244cdb058d8312bcf3909 : SUCCESS in 7m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/744d1ea4764546579c9ec59b1fdb0be4 : FAILURE in 36m 34s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8070a805c5024710a62c8a74b01d08f3 : SUCCESS in 1h 15m 50s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/d807b42911c84e2c8b8eee4da693b7f2 : SUCCESS in 2h 09m 30s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/e1abc134183342e7be6bd1e8cbfdb7fc : SUCCESS in 1h 43m 40s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/97c00caca5524c74af4b8304f4d64b1a : SUCCESS in 2h 02m 41s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/dc3dd0589a73496984675e4cd7da22f1 : FAILURE in 42m 40s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/7530f91580324498ae6d046158292f78 : FAILURE in 28m 28s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3330db76864b4a10aa1ddb4c9d4b22fd : FAILURE in 1h 04m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ef56c03cde54a28b289ba454be698be : FAILURE in 43m 57s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/5ee596791e2d4d61b77d97061de5b70c : SUCCESS in 1h 50m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3c14b77b6e5a80be0dd0be5eda3ceb21cd1c55f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 05:04:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ad1a1578c342f176e6efd3589e09a9897a19f26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:18:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14f6ddc86bc447708495c68deee91489 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f3ba7aeb9b41739eded516ac05ff61 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/228231bc8d924c54927bd6d5c1ade3c0 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1fc359f8e564cf8ae22148f29d673ce : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b84f6b952824cac9658a7f2c7068139 : SUCCESS in 5m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cab7ea4cac6a4ee7bf57ed8431943493 : SUCCESS in 7m 56s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2ba20297820f41899a857a9de68b5b49 : SUCCESS in 1h 20m 13s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ec555e597d8948a0a6998a22046d20b6 : SUCCESS in 2h 13m 22s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c26158aee0fb4227aa74e1144e2b7f14 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9be83a902bbd5ad4c8cf0f71010116f78fa0cb24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:18:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8f91637b1c432553d582eb42eb1dc79c0cbaf2d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:21:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22597821b274744b981b316331c4941 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5d78f68925b0f73e85c39308439eec0f20aad9f","revisions":{"c5d78f68925b0f73e85c39308439eec0f20aad9f":{"kind":"REWORK","_number":1,"created":"2020-06-15 22:30:41.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/43/735743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/43/735743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/735743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/735743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/43/735743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/43/735743/1"}}},"commit":{"parents":[{"commit":"775abab528f7505209f20dd184b7ba18be7f9351","subject":"Support image type for guest image building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/775abab528f7505209f20dd184b7ba18be7f9351"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-15 22:30:17.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-06-15 22:30:17.000000000","tz":720},"subject":"Fix building raw image","message":"Fix building raw image\n\nChange-Id: Ie49b0622d8ae6b9b7189173a02f21a28c8dfca04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5d78f68925b0f73e85c39308439eec0f20aad9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5d78f68925b0f73e85c39308439eec0f20aad9f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
