)]}'
{"id":"openstack%2Ftrove~738332","triplet_id":"openstack%2Ftrove~stable%2Ftrain~Ifb12da84b343942b516416cda0e644d095964745","project":"openstack/trove","branch":"stable/train","topic":"doc-train","hashtags":[],"change_id":"Ifb12da84b343942b516416cda0e644d095964745","subject":"[docs] Update building_guest_images for train","status":"ABANDONED","created":"2020-06-28 02:18:48.000000000","updated":"2020-08-04 06:47:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a77e134ab4f7ec733d3e071259848a54cd7b6005","_number":738332,"virtual_id_number":738332,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-28 02:33:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-28 02:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e840d0b0848d2d5f0cab4400c584d1382c14192c","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-28 02:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82a0f14de8640583fd2e7062d08890d8d9a9b8ad","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-06-28 02:22:07.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8bdb687616a47f403a8180368f503aeca8e6c443","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 02:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709550bf69c44f1b939d8b0666b5e9ab : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63eb15b4188643c587487360fd8db4fa : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"ff13569fe59d425eb2b84f78c9e919a8ac8ebe1e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-08-04 06:47:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4911d78ff5142356e8984a7dfcd87b46ea727f22","revisions":{"87af3935ae1b17a95733c899412e9638021832ac":{"kind":"REWORK","_number":1,"created":"2020-06-28 02:18:48.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/32/738332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/738332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/738332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/738332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/738332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/738332/1"}}},"commit":{"parents":[{"commit":"d8b5525fd2a544c264a86edcd0d3493709f262af","subject":"Fix CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8b5525fd2a544c264a86edcd0d3493709f262af"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-06-28 02:09:35.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-06-28 02:09:35.000000000","tz":480},"subject":"[docs] Update building_guest_images for train","message":"[docs] Update building_guest_images for train\n\nupdate trovestack document url from latert to train, old url cannot be accessed\n\nChange-Id: Ifb12da84b343942b516416cda0e644d095964745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87af3935ae1b17a95733c899412e9638021832ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/87af3935ae1b17a95733c899412e9638021832ac"}]},"branch":"refs/heads/stable/train"},"4911d78ff5142356e8984a7dfcd87b46ea727f22":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-28 02:22:07.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/32/738332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/738332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/738332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/738332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/738332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/738332/2"}}},"commit":{"parents":[{"commit":"d8b5525fd2a544c264a86edcd0d3493709f262af","subject":"Fix CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8b5525fd2a544c264a86edcd0d3493709f262af"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-06-28 02:09:35.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-06-28 02:19:38.000000000","tz":480},"subject":"[docs] Update building_guest_images for train","message":"[docs] Update building_guest_images for train\n\nUpdate trovestack document url from latest to train,\nold url cannot be accessed\n\nChange-Id: Ifb12da84b343942b516416cda0e644d095964745\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4911d78ff5142356e8984a7dfcd87b46ea727f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4911d78ff5142356e8984a7dfcd87b46ea727f22"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
