)]}'
{"id":"openstack%2Ftrove~741084","triplet_id":"openstack%2Ftrove~stable%2Focata~I8cd5bc59bc64ed170c65c7d905ffa06c856f666a","project":"openstack/trove","branch":"stable/ocata","topic":"ocata","hashtags":[],"change_id":"I8cd5bc59bc64ed170c65c7d905ffa06c856f666a","subject":"ignore_dbs add sys for mysql57","status":"ABANDONED","created":"2020-07-15 05:11:35.000000000","updated":"2020-08-11 03:46:20.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4a20f95869bbb68849973de12cad637019f0c7d4","_number":741084,"virtual_id_number":741084,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-21 02:50:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 02:50:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fc8e816f30bf6e467943796cab248728bfb8ec5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-15 05:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"771c2d95f4c7aae397bf952d6a4eddbcea7eb028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 05:30:24.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5a2029b0694042b0c52ca218a34bf4 : FAILURE in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6314639e4234847abff2662caacefa6 : SUCCESS in 10m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2a5927863d604ec382593f07e90e6f01 : SUCCESS in 17m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a7f2846330584fd29d56064156519c57 : SUCCESS in 5m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8d707de4d5094e99b47f3307b0c3f17a : FAILURE in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6b9c0f22961e8d2b58c1421f971cb8db7f344535","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-15 06:01:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4e12fb470e68a1b9f4ee1e0bd0dc7be17cccc6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 06:30:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cd5c454b9124fd8bc8c7bc6bdc86f10 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d008576af6fd4d6bbc969cc0820e85e8 : SUCCESS in 13m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/203c54f936c343289c79d9ee5a591716 : SUCCESS in 17m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/64ad58e9235e4db4b815d0c33d3ca18d : SUCCESS in 6m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/643b835313734215a78ec150e2e42987 : FAILURE in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"405b6153b61e884ea12271ecd014ee9172d85503","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-21 02:12:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d6f58df7556d95be28c02fd784c290c211186682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 02:25:30.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f2a6d1c7834691ba2031bcc418137c : FAILURE in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0deef18a038b48f49a7ca000834cdd5e : SUCCESS in 9m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b79773c270ef49528a2f45adfbac3032 : SUCCESS in 11m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/62016d4aee4c4dffac0703e58ba8a173 : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c1afb243fd1640b2bd389c01fdb41408 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":3},{"id":"dfca137c7ecc8f65615c06dcc57969cab150c11d","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-07-21 02:33:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"818f952ccf3a799cbc31f3b206e8b47980f303e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 02:50:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b97e48735374c8792c0157a2c5e51a3 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/986dbe4af0b54b689deec7e7090cb15e : SUCCESS in 9m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1412166ed6714698b8ddda39d41c1d78 : SUCCESS in 11m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/102c030245d34fb89b0770e44f9eb3d1 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b7198c715ebb4f5e95fc75f23063e13e : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":4},{"id":"ebe5663f66f1ed00d262c4e863c622e1e41b3344","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-08-11 03:46:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7c0208c3347636038b6c265d0cd3967e5ef3335f","revisions":{"562b60e998e4e2165d3dcb492618598e17b8775d":{"kind":"REWORK","_number":1,"created":"2020-07-15 05:11:35.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/84/741084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/741084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/741084/1"}}},"commit":{"parents":[{"commit":"12ba44ad3654a51c41771cfd7c3491539e0fcf31","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba44ad3654a51c41771cfd7c3491539e0fcf31"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-07-15 05:06:50.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-07-15 05:09:20.000000000","tz":480},"subject":"ignore_dbs add sys for mysql57","message":"ignore_dbs add sys for mysql57\n\nUpdate ignore_dbs argument for mysql57 for ocata\n\nChange-Id: I8cd5bc59bc64ed170c65c7d905ffa06c856f666a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/562b60e998e4e2165d3dcb492618598e17b8775d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/562b60e998e4e2165d3dcb492618598e17b8775d"}]},"branch":"refs/heads/stable/ocata"},"00737f375cf002803697e8b9d5e4007118432137":{"kind":"REWORK","_number":2,"created":"2020-07-15 06:01:34.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/84/741084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/741084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/741084/2"}}},"commit":{"parents":[{"commit":"12ba44ad3654a51c41771cfd7c3491539e0fcf31","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba44ad3654a51c41771cfd7c3491539e0fcf31"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-07-15 05:06:50.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-07-15 05:59:01.000000000","tz":480},"subject":"ignore_dbs add sys for mysql57","message":"ignore_dbs add sys for mysql57\n\nUpdate ignore_dbs argument for mysql57 for ocata\n\nChange-Id: I8cd5bc59bc64ed170c65c7d905ffa06c856f666a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00737f375cf002803697e8b9d5e4007118432137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/00737f375cf002803697e8b9d5e4007118432137"}]},"branch":"refs/heads/stable/ocata"},"f9490da7f8a9e2627fc2b4241e5cd0af1792d336":{"kind":"REWORK","_number":3,"created":"2020-07-21 02:12:05.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/84/741084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/741084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/741084/3"}}},"commit":{"parents":[{"commit":"12ba44ad3654a51c41771cfd7c3491539e0fcf31","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba44ad3654a51c41771cfd7c3491539e0fcf31"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-07-15 05:06:50.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-07-21 02:10:25.000000000","tz":480},"subject":"ignore_dbs add sys for mysql57","message":"ignore_dbs add sys for mysql57\n\nUpdate ignore_dbs argument for mysql57 for ocata\n\nChange-Id: I8cd5bc59bc64ed170c65c7d905ffa06c856f666a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9490da7f8a9e2627fc2b4241e5cd0af1792d336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9490da7f8a9e2627fc2b4241e5cd0af1792d336"}]},"branch":"refs/heads/stable/ocata"},"7c0208c3347636038b6c265d0cd3967e5ef3335f":{"kind":"REWORK","_number":4,"created":"2020-07-21 02:33:21.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/84/741084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/741084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/741084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/741084/4"}}},"commit":{"parents":[{"commit":"12ba44ad3654a51c41771cfd7c3491539e0fcf31","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/12ba44ad3654a51c41771cfd7c3491539e0fcf31"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-07-15 05:06:50.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-07-21 02:31:21.000000000","tz":480},"subject":"ignore_dbs add sys for mysql57","message":"ignore_dbs add sys for mysql57\n\nUpdate ignore_dbs argument for mysql57 for ocata\n\nChange-Id: I8cd5bc59bc64ed170c65c7d905ffa06c856f666a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c0208c3347636038b6c265d0cd3967e5ef3335f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7c0208c3347636038b6c265d0cd3967e5ef3335f"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
