)]}'
{"id":"openstack%2Ftrove~741334","triplet_id":"openstack%2Ftrove~master~I0ddd7214dae6e29ddfaf045fdb282f4980a8afff","project":"openstack/trove","branch":"master","topic":"container-name","hashtags":[],"change_id":"I0ddd7214dae6e29ddfaf045fdb282f4980a8afff","subject":"Support backup strategy API","status":"MERGED","created":"2020-07-15 22:31:49.000000000","updated":"2020-07-17 07:16:36.000000000","submitted":"2020-07-17 07:12:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741334-1594969920923-78954b85","meta_rev_id":"816bdad549cf18147048e6e56aed07195a717079","_number":741334,"virtual_id_number":741334,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-17 07:12:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-07-17 04:50:10.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-07-17 04:50:10.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 07:12:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36c01a163f07eb143f0f6a6941c94c32961d108d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-15 22:31:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1158f2970d557b949940dca00caa8d905289a707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:51:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a9ad234c01d4395818a35c06bfcf2df : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ba6469515774ff4af0b0ffb8a1fb36d : FAILURE in 7m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3520077163df40ff8376969e032dd5e1 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bc138682ab47c0b90785341e051105 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa92c46c7d244c86a003cc718d81c0c0 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9baa7d3fcb194fe59f8a3c01e9b81eb1 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdba1edc31e443ebbdd9573a5fcfb919 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c8820569b6d40c6a96b5b0fc1255c54 : SUCCESS in 6m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ed91bda8f634985b9ed56c92b86930d : SUCCESS in 6m 34s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/883413b7293d421c83ac88e919d9079b : FAILURE in 2h 11m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e49a4813655645458748ec9a1cf115f4 : FAILURE in 1h 21m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0be32d092255eb349fdfdad4baa14001661d8e3c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-16 11:04:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9563b34f8c461e9eb6d7574d6fe66c0a565da53d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 13:21:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d16becf04ca4b7483264eaedea47766 : SUCCESS in 5m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e16d033a685e43b5ace9816ae838a6fc : FAILURE in 7m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/258179cc57514309ad39523b99eeada7 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a120e75351244da6b6f6b50368bab4d1 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f33010f3edbd4182b4a45ba84cabde3a : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b0f886707744299b72f40ec9938186b : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8604870972c64665a5b431b629c301d7 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8295d4903fc043e5b1d14337de005a48 : SUCCESS in 5m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1188cefd3fe14aeaac954a899b81d28e : SUCCESS in 9m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b6f3262dd1864b32940602dbf34ea5be : FAILURE in 2h 15m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61061a5df69445e28cb02c2c2e4b120b : SUCCESS in 1h 05m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a8ad21f41f968c42b3367ea78efe081d2c8b4b4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-16 23:36:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bfc82cc4cc105ab48f19a4509220a4dc1ad4d1f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 01:40:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/018fb3bc8a014529bebb11a608c82677 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bef86b1e4d9b46108c65167096e13ae1 : FAILURE in 7m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b99e4365d4de4ea8af38a68849d440bc : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2527a36261114d098bf1cc45b42e102d : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30cf21848dff46e39e3ff265e9292571 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01da472af4404705b8d42d28002a8730 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865b2550c0b14feca316800d8707fe1c : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37e0a7bafdf4443384b3148bd3f0f263 : SUCCESS in 6m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2641281f0b04533928ed420b72ab59f : SUCCESS in 7m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e0085c258f3c49f5bef9ff4ac8692f1f : FAILURE in 2h 02m 32s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e9110e0eb64dd49957654e9035c293 : FAILURE in 1h 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8e8d4a2e1494ad8180f2f39b82985270814279f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-17 04:50:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"82d713251c2f509ff0782af5d84604773eb7ddbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 04:50:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2d8f74c0623d7eea6726351c9094cb6d733150c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 07:12:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cb0cc12b294cc05b6f3e057626f48e029ed328a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 07:12:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b80a3f293f0148fc8ab59be9b8a6ed9e : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ed659e1fb2345d08827f6ed2c1389e8 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e697131c9a49e6bea8c8cdf0daadf1 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2abf288ac594f6eb9109ded2194892e : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2876b72c82d74861a60219c92c0c4ed0 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bfae4fbd7a540668f59fd59fb022190 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce1b2ae5d2443818ac905369b1f9af8 : SUCCESS in 7m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6073fd9bb4aa4c278ae7843817aa073b : FAILURE in 2h 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"933790ff06e64271ca436a252b49ee66314652e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 07:16:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b90d208b2d7e4b6098ec68295790125a : SUCCESS in 2m 29s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9652a9b515a4ed89876e2623c1b3c54 : SUCCESS in 2m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddda433ac9b74ca1a94531f51d33bcf7 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"828e873846495df6d6911f9c196ae1fb478492f5","revisions":{"5b28fe3a74099401da1060b894820c7b1e6c5413":{"kind":"REWORK","_number":1,"created":"2020-07-15 22:31:49.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/34/741334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/741334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/741334/1"}}},"commit":{"parents":[{"commit":"f8ca333b4302ef5681f7f63731cce141e4ad7cf3","subject":"Correct some information in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8ca333b4302ef5681f7f63731cce141e4ad7cf3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-12 09:26:44.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-15 11:06:17.000000000","tz":720},"subject":"Support swift container name for creating backup","message":"Support swift container name for creating backup\n\nChange-Id: I0ddd7214dae6e29ddfaf045fdb282f4980a8afff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b28fe3a74099401da1060b894820c7b1e6c5413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5b28fe3a74099401da1060b894820c7b1e6c5413"}]},"branch":"refs/heads/master"},"6c8758afae0e305e2af11425fbaabac4b570b4e0":{"kind":"REWORK","_number":2,"created":"2020-07-16 11:04:57.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/34/741334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/741334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/741334/2"}}},"commit":{"parents":[{"commit":"f8ca333b4302ef5681f7f63731cce141e4ad7cf3","subject":"Correct some information in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8ca333b4302ef5681f7f63731cce141e4ad7cf3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-12 09:26:44.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-16 11:04:07.000000000","tz":720},"subject":"Support backup strategy API","message":"Support backup strategy API\n\nChange-Id: I0ddd7214dae6e29ddfaf045fdb282f4980a8afff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c8758afae0e305e2af11425fbaabac4b570b4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6c8758afae0e305e2af11425fbaabac4b570b4e0"}]},"branch":"refs/heads/master"},"828e873846495df6d6911f9c196ae1fb478492f5":{"kind":"REWORK","_number":3,"created":"2020-07-16 23:36:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/34/741334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/34/741334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/34/741334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/34/741334/3"}}},"commit":{"parents":[{"commit":"f8ca333b4302ef5681f7f63731cce141e4ad7cf3","subject":"Correct some information in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f8ca333b4302ef5681f7f63731cce141e4ad7cf3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-12 09:26:44.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-16 23:35:53.000000000","tz":720},"subject":"Support backup strategy API","message":"Support backup strategy API\n\nChange-Id: I0ddd7214dae6e29ddfaf045fdb282f4980a8afff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/828e873846495df6d6911f9c196ae1fb478492f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/828e873846495df6d6911f9c196ae1fb478492f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
