)]}'
{"id":"openstack%2Ftrove~741806","triplet_id":"openstack%2Ftrove~master~I3398557700dd76ec831e441ba074032fd4b67cca","project":"openstack/trove","branch":"master","topic":"fix-slow-start","hashtags":[],"change_id":"I3398557700dd76ec831e441ba074032fd4b67cca","subject":"Fix the slow start of mariadb container","status":"MERGED","created":"2020-07-18 10:06:35.000000000","updated":"2020-07-18 12:10:34.000000000","submitted":"2020-07-18 12:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741806-1595074120528-45274524","meta_rev_id":"3c308c374b31113328dd8a6945ea31e0c6ab00e6","_number":741806,"virtual_id_number":741806,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 12:08:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-07-18 11:29:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-07-18 11:29:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-18 12:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab265a9fae1f8d80607c6338951c887b58119649","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-18 10:06:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45cf18ff90ad805aa61643b22c5b00039f4ac69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 10:40:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb6285537d104f718ccde3d962f26ebb : FAILURE in 6m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28bf8cc1b41d45bc93b41e17ebd2846d : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881de4bc95c942299bd0d1bfa8f1d25b : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b24af1e1afc4d638042714b6cb3dc6a : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40bf8250462f44f3964b4c06980b087c : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/480b22517fea4431b49d63f897d05daa : SUCCESS in 5m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/396539c3e86d4ab197612454db2cb537 : SUCCESS in 5m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/299a4aac692c48e4b6ab0a81529aa9f7 : FAILURE in 28m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/222194058be94964a4f21fb4c8141a64 : FAILURE in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0528f060c63aced6d2b37e3b88a3b9163ff4da1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-18 11:29:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a30521418b1a6b21d38c3af7ce8456aa22e6ab24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 11:29:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98c3bd7ea8145cef201e5ebf2a38194f2df979f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 12:08:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3fc74015acb4cb288fae77db0b988a5 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9990be772dc0457ebf273ab735ed992e : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7430ee5080284a3688583c8aacdade2e : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7ae643e6a174873b3076f62ff683bb7 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de393b25d304df5be6cb3bfff324967 : SUCCESS in 6m 47s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c1818eb210e547d09df28cb98dd9c92e : FAILURE in 23m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bec141e0811f1753b2ab8e2c1379450441a75b69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 12:08:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e62dd27b40cebed79309afe1d2abf68a92206e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 12:10:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62085b994924bc2b2cae842aa5d8838 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"058490a609108bf0709117d42710ad3f7ee28d70","revisions":{"058490a609108bf0709117d42710ad3f7ee28d70":{"kind":"REWORK","_number":1,"created":"2020-07-18 10:06:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/06/741806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/06/741806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/741806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/741806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/06/741806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/06/741806/1"}}},"commit":{"parents":[{"commit":"828e873846495df6d6911f9c196ae1fb478492f5","subject":"Support backup strategy API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/828e873846495df6d6911f9c196ae1fb478492f5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-18 09:55:20.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-18 10:00:42.000000000","tz":720},"subject":"Fix the slow start of mariadb container","message":"Fix the slow start of mariadb container\n\nAdding \u0027-e MYSQL_INITDB_SKIP_TZINFO\u003d1\u0027 when starting database\ncontainer.\n\nRef:\nhttps://github.com/docker-library/mariadb/issues/262\n\nChange-Id: I3398557700dd76ec831e441ba074032fd4b67cca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/058490a609108bf0709117d42710ad3f7ee28d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/058490a609108bf0709117d42710ad3f7ee28d70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
