)]}'
{"id":"openstack%2Ftrove~742577","triplet_id":"openstack%2Ftrove~master~I000a4e90800454972dd39f2f82d286571bc0b96c","project":"openstack/trove","branch":"master","topic":"online-resize-volume","hashtags":[],"change_id":"I000a4e90800454972dd39f2f82d286571bc0b96c","subject":"Support online volume resize","status":"MERGED","created":"2020-07-23 02:30:59.000000000","updated":"2020-07-25 11:13:25.000000000","submitted":"2020-07-25 11:12:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742577-1595675537580-85cd6830","meta_rev_id":"79913cc0a86c89666309b14bcda92c4e60db1ea8","_number":742577,"virtual_id_number":742577,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 11:12:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-25 08:55:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-25 08:55:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 07:41:51.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-07-25 11:12:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4feed13033c367d2c881ec120e862ddcd89eafde","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-23 02:30:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6650d862a8ab96296703d4c42e24fa2b0e7b1536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 04:45:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27b88f488b8745bfb8b00c110d7d4dcb : FAILURE in 6m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375e09043e9241288a22af599b1508af : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4c7d497b504e908d12efe47566237f : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42d72e41d22645f5afa83c26ed6bb8fa : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9376aad3b0b4737b9e3bfb9dc0a1f4b : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa5f83aa53d4fcc902f715a4ecc47fd : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784951ce3c7e4582bf440d7d1c755ddf : SUCCESS in 5m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76c64af615904a64956bf45e67783fff : SUCCESS in 5m 34s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/fe47ba530f154e68ba00b5f360577a6e : TIMED_OUT in 2h 12m 54s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51cec970ef514fd7ae7dabe0fec04b0a : FAILURE in 1h 06m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"437968e144215b859f1983703e3708c1a4bcef0f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-07-23 07:41:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5d6424fcc6f2e67133f8ac277bcdb584ad6d092","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-24 02:08:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"104aba013c2ed1c56eb08d655bec53e8e78b6358","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-24 02:44:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a809183be3db914d6ccde02e9bc63db07ac7080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 04:27:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05e17945827a49eb81228382bdbb1a34 : FAILURE in 6m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/084cb0d6ae494e49bac1b85507a1f30c : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda79df9cad948eb950dd37fb13abe7f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbecc9d21d6549459fb2c5e743ced0d9 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c40e5489c177497797947ddbf069917a : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d921d4f90c454d5ab2728fa0549880dc : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/897e4926fc2843c6b185bec462584a1e : SUCCESS in 8m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c3273223aa71415d82613bdb12917c03 : SUCCESS in 8m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/25e933aff9344eb5939ab3461db7f5c0 : TIMED_OUT in 2h 12m 49s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3e9d5934e947dd9b09f79ea581ddd2 : FAILURE in 1h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64cec1a92da33b401a5a9678384cdcf507c0c9cd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-25 08:55:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf5c39c88ecec5f083ffc5496433aabd14c22e34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 08:55:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5945243dfbb8067cfd15eb64cb370db704f213e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:12:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"022b57f35e96f41128044f601e9b454c809e7c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:12:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75c1947398e54bd3a179eb1f10bf9562 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3105b9d6822544cc96e5149c85b912c7 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e0b57f8ac434de0b4b9f9e1d69a2b31 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6f7cfde5c3a4861a7bf5dda32442c8b : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7ac1a2fafd4ddc95a1036fdefd5ff4 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d77e5029b354e3bb2c5eec91000f2ff : SUCCESS in 7m 40s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/98511bfa9bcf44efa0c64ad2d94b2e3b : TIMED_OUT in 2h 15m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e0476d9b30ee5dbbbbc7051671511da0cb3b5ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 11:13:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4ce613cd0346f290fb49727be7e23a : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c1f53ab2d4d4eadbf9688150f91e298 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba046b2a14459132ca10d30a1d2cc0a01126bb1d","revisions":{"ee6fead9dc0a605f28f786c0954034efb864e808":{"kind":"REWORK","_number":1,"created":"2020-07-23 02:30:59.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/77/742577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/742577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/742577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/742577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/742577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/742577/1"}}},"commit":{"parents":[{"commit":"39b0df0a6b87a6f2cb748cb81ac376088f34b88f","subject":"Merge \"Fix the slow start of mariadb container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39b0df0a6b87a6f2cb748cb81ac376088f34b88f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-22 03:41:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-23 02:29:48.000000000","tz":720},"subject":"Support online volume resize","message":"Support online volume resize\n\nChange-Id: I000a4e90800454972dd39f2f82d286571bc0b96c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee6fead9dc0a605f28f786c0954034efb864e808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ee6fead9dc0a605f28f786c0954034efb864e808"}]},"branch":"refs/heads/master"},"ba046b2a14459132ca10d30a1d2cc0a01126bb1d":{"kind":"REWORK","_number":2,"created":"2020-07-24 02:08:06.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/77/742577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/742577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/742577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/742577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/742577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/742577/2"}}},"commit":{"parents":[{"commit":"39b0df0a6b87a6f2cb748cb81ac376088f34b88f","subject":"Merge \"Fix the slow start of mariadb container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39b0df0a6b87a6f2cb748cb81ac376088f34b88f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-22 03:41:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-24 01:55:51.000000000","tz":720},"subject":"Support online volume resize","message":"Support online volume resize\n\nTrove now supports to resize volume without downtime. To use this\nfeature, the version of Nova and Cinder needs to be at least Pike, the\nconfig option ``cinder_service_type`` needs to be set to ``volumev3``.\nThe cloud admin can disable this feature by setting\n``online_volume_resize\u003dFalse``, default is enabled.\n\nChange-Id: I000a4e90800454972dd39f2f82d286571bc0b96c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba046b2a14459132ca10d30a1d2cc0a01126bb1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba046b2a14459132ca10d30a1d2cc0a01126bb1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
