)]}'
{"id":"openstack%2Ftrove~743093","triplet_id":"openstack%2Ftrove~master~Icadc95ea54e4509dc148f8e84f2eaac5840509f3","project":"openstack/trove","branch":"master","topic":"replica-same-size","hashtags":[],"change_id":"Icadc95ea54e4509dc148f8e84f2eaac5840509f3","subject":"Using same config with primary for replicas","status":"MERGED","created":"2020-07-26 23:25:21.000000000","updated":"2020-07-27 04:09:11.000000000","submitted":"2020-07-27 04:08:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743093-1595822887126-d1a6a7d1","meta_rev_id":"cda851b445e56304c3621ba7d8d3cfeba3b1cc62","_number":743093,"virtual_id_number":743093,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 04:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-07-27 02:12:19.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-07-27 02:12:19.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 04:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4bd3fdd910942736a2756f56fef6966f894c3d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-26 23:25:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56c78716cea27f169ae4dae59f8c6c952ece9476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 01:23:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d20a17802d94280b49770d9443d1472 : FAILURE in 8m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bed269ff1e04c62ba0b14d62389c584 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce470864a2a44896aa4786ba01d6967a : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aff427f4ce64aa9be9714a5be8e5db6 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28883360b9424fd2af3c729b7ddceea1 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8db407998fe45f6b4b1173f73cdc5de : SUCCESS in 7m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8678efb425f74e67a768f4f02ccf701f : SUCCESS in 7m 56s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8c379a28c4124251bb7d5bb80085ae35 : FAILURE in 1h 53m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d060b432a8914cdb996cc9ddf408573c : FAILURE in 1h 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6484c8d126274078ffad08166ed4c719bc0695d8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-07-27 02:12:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c246cec4e1967bc32dcd3bf7376c4c539b35dfcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 02:12:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b4e0c3c0bae2de516d35bc1cf832ccf9d268d45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 04:08:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ead6c1e83a949908054f49b90abedce : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f1f1c2c1f9458183aa01a6cb1882d9 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/222ebd0a44f540c39a9e2c28bca91952 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2afebf66e2f4ec089655e36f90b0d3d : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf5a9ac5c984c2cbc5097b0da9d432c : SUCCESS in 6m 38s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1e637c8647cc4b24945e60b9f7a08206 : FAILURE in 1h 54m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a281cabcfc9bb9e9b0b26ae7e99bdb90f8e8c97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 04:08:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fed49e6bc40372964ce6ed5f594cd7652ec267b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 04:09:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a747e65a75164938ade9c03f15fc6233 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efb6a811bedd07819d95ff462195929dd1a33922","revisions":{"efb6a811bedd07819d95ff462195929dd1a33922":{"kind":"REWORK","_number":1,"created":"2020-07-26 23:25:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/93/743093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/93/743093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/743093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/743093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/743093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/93/743093/1"}}},"commit":{"parents":[{"commit":"39b0df0a6b87a6f2cb748cb81ac376088f34b88f","subject":"Merge \"Fix the slow start of mariadb container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/39b0df0a6b87a6f2cb748cb81ac376088f34b88f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-25 08:39:47.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-07-26 21:28:17.000000000","tz":720},"subject":"Using same config with primary for replicas","message":"Using same config with primary for replicas\n\nChange-Id: Icadc95ea54e4509dc148f8e84f2eaac5840509f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efb6a811bedd07819d95ff462195929dd1a33922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/efb6a811bedd07819d95ff462195929dd1a33922"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
