)]}'
{"id":"openstack%2Ftrove~744371","triplet_id":"openstack%2Ftrove~stable%2Fussuri~I44e134efbd5939962a508468584defcf4cb4c0c5","project":"openstack/trove","branch":"stable/ussuri","topic":"ussuri-replication-floatingip","hashtags":[],"change_id":"I44e134efbd5939962a508468584defcf4cb4c0c5","subject":"Do not operate floating IP for relication cluster","status":"MERGED","created":"2020-08-02 09:50:27.000000000","updated":"2020-08-03 23:26:57.000000000","submitted":"2020-08-03 23:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744371-1596496829724-7b2e085e","meta_rev_id":"aabe4accb03bbc1ae1255cbd564398d60e1c8522","_number":744371,"virtual_id_number":744371,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 23:20:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-08-03 21:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-08-03 21:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 23:20:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02e8c430162d80f5c0ed51198ee22576e2c2d634","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-02 09:50:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd0c5aff97ebd0e25de68f991164249b0b1c7d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 11:52:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c419fdbc264832ad14cbb2547fcfdb : FAILURE in 12m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f2e9728bec54325953ea5f71a75ebe9 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6534e919188740ddb80ac752dac57058 : FAILURE in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73daf18cfde48598475b7a88fb59a0a : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd6d2d68f94941be97f149462bfe90cd : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2bd16cae96e4cf08eaa16601a6ac746 : SUCCESS in 4m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09b3bbf5112c4f0cb8e7c3556994355d : FAILURE in 5m 13s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a7d1f98576e340c488c5558eef98e665 : SUCCESS in 6m 24s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/af28baada4ea43c689dece781d81253c : FAILURE in 38m 17s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/42069ee031594bffaf4f3320f5c569af : SUCCESS in 1h 47m 05s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b3ec94a5e37b4914910de369ceb912ca : SUCCESS in 2h 00m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/f743c9f6593849389427df897db79bc6 : SUCCESS in 1h 32m 56s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/2d62d9df7dba4e32901efbc8e1071732 : FAILURE in 1h 47m 35s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/0e1c57335cee4d1b9bae4b079e98760e : FAILURE in 47m 50s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/4607ed8c53c04d54a01657c6de53e69b : FAILURE in 27m 20s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b3dd12217b9c4e4eb6a1c52cf9f3f572 : FAILURE in 39m 46s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15eb259ea03b4fc5b17188c89bda6114 : FAILURE in 37m 09s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/8f792d4d75204d01afa07d304947f544 : SUCCESS in 1h 51m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e82aacd7d3b82059360502abdb80b17de80186b0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-03 09:12:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e744c5477c1c3f6ff41ffac54d1e5c4f62646c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 11:40:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17ed5389b6e445cca56b35f693ad1f4a : FAILURE in 10m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fdcbbfb3cda415185e01ca12d023f8f : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b97588967b846eb94fafe2cbcb5be4c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8959f05e5e2b4d4ca023788c91d95b7b : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/653868a2d2d14f1ea64aaa773415490f : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148a4ebbaed14dcfa362dffda8f228cd : SUCCESS in 5m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1202f3e793d4143a3b6ee76b0be1bf9 : SUCCESS in 5m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c41a3f9994f5414ab149eaeb49f1bf90 : SUCCESS in 59m 57s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6b8165c8c4b94ee39ca9bcc951e737fe : SUCCESS in 2h 03m 19s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/91c603b6d2eb4a8aaf4eb73eb30e0eb1 : SUCCESS in 2h 18m 56s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/c3a5a44b817f498bae93b2773a5a0bc7 : SUCCESS in 1h 28m 41s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/bb1521e579204161ae83b66ba5a29440 : SUCCESS in 2h 13m 15s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/1a667ad177dc46f3b6f33cf80218ae38 : FAILURE in 49m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/97284f7787014494a0c43d98f85fb778 : FAILURE in 44m 04s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/f739221f5e2744658aafa43fcd4e8ea8 : FAILURE in 1h 02m 31s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df326e106efc4e28847df982fbeaad76 : FAILURE in 27m 18s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/6320c0db4a8f4b0882710c9b3a5c4d34 : FAILURE in 1h 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f00a667c1e6f56d495abce1d3d7e2989dda0a27e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-03 21:01:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4fb6d6876477eec7becd1dd3472318d646831215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:01:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33ac360a551de57e4c52453078ea00d6639159f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:20:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ac0783ec93a6f306146de13153a2cdd8cdbcb4d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:20:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c10f553b0a84298baa0e8dd115bcce6 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9a78f811683453e9110a0e42e44c28d : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dcaa924b0614d92b23e11bd05987885 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d023270b386f464cb974e40a5e28ef00 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723822b4da294b489ad89df1e5d39c93 : SUCCESS in 8m 04s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/df836490c03d4f50ac33446b9c284f7f : SUCCESS in 2h 16m 49s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/c81c2840f27a4b1cb4c1c65a6d094582 : SUCCESS in 1h 28m 53s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/0a0cb63f2c1c4c4f98ae52ca02bbfe1d : SUCCESS in 1h 54m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7f9d2fa48ed900bd9c1cbd23ea4109aa3f380e6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 23:26:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0cf54623a9413d8d94afac8b926a4b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"695a63e228548349cfc8d7ced75ecc53e1b1f984","revisions":{"d8f143b5e17b799402082ea2f2b10cfdd1ee287d":{"kind":"REWORK","_number":1,"created":"2020-08-02 09:50:27.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/71/744371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/744371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/744371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/744371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/744371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/744371/1"}}},"commit":{"parents":[{"commit":"c5d78f68925b0f73e85c39308439eec0f20aad9f","subject":"Fix building raw image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5d78f68925b0f73e85c39308439eec0f20aad9f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-02 09:44:48.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-02 09:44:48.000000000","tz":720},"subject":"Do not operate floating IP for relication cluster","message":"Do not operate floating IP for relication cluster\n\nChange-Id: I44e134efbd5939962a508468584defcf4cb4c0c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8f143b5e17b799402082ea2f2b10cfdd1ee287d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d8f143b5e17b799402082ea2f2b10cfdd1ee287d"}]},"branch":"refs/heads/stable/ussuri"},"695a63e228548349cfc8d7ced75ecc53e1b1f984":{"kind":"REWORK","_number":2,"created":"2020-08-03 09:12:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/71/744371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/71/744371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/744371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/744371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/71/744371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/71/744371/2"}}},"commit":{"parents":[{"commit":"c5d78f68925b0f73e85c39308439eec0f20aad9f","subject":"Fix building raw image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c5d78f68925b0f73e85c39308439eec0f20aad9f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-02 09:44:48.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-03 09:11:53.000000000","tz":720},"subject":"Do not operate floating IP for relication cluster","message":"Do not operate floating IP for relication cluster\n\nChange-Id: I44e134efbd5939962a508468584defcf4cb4c0c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/695a63e228548349cfc8d7ced75ecc53e1b1f984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/695a63e228548349cfc8d7ced75ecc53e1b1f984"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
