)]}'
{"id":"openstack%2Ftrove~744619","triplet_id":"openstack%2Ftrove~master~Idcaab182a1b42b2506dedad4f532dae05b485cd9","project":"openstack/trove","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Idcaab182a1b42b2506dedad4f532dae05b485cd9","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-04 03:32:55.000000000","updated":"2020-09-08 23:54:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d9b35e127e4ecab97412a39113316392bf6b6b0","_number":744619,"virtual_id_number":744619,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-08 23:53:50.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-04 05:29:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 05:29:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7289f39124ccf3c3d32efc67c7e00887a736118f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 03:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb86d8c8fdafe881873e56ea6ac061c2aba3c715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 05:29:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/219a940c75a344a5b604027bb0b4638e : FAILURE in 6m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37fb71d5a8c943f2bc2117d375f5ad87 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860309e00b2f41a696e7a1f97923a891 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4aa4cad5af94dc89d7fb2d28733b341 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59fe6cf079a14884ad838246adae4622 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a8525e2cbf4145ba749f80bdcd43a2 : SUCCESS in 6m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8f930ac1ec6f479eb21e1f1172ef6b4c : SUCCESS in 5m 28s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/43ff3665646d4c6b9af4de005349a175 : FAILURE in 30m 52s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fe9f4f5407d4155a67497e60e75e825 : FAILURE in 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82d8757328a5e702eb5e4129af9cddc20b86d799","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-04 21:08:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2caab0b83a64c33cd7492b4354807d312ffdad9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 23:40:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85fcd7f87b694480870bebaefdcae8a9 : FAILURE in 8m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b08334ecc4c84eaea015553a475be234 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a46848b9f834a5e9ec2beae9deb32cf : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5edee9206e4945c89393aa5ee5aedd5b : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdb46f22603e46fea8fd86e3c4b6deb9 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89de57f462f44fafac7eba7b4c2b47dd : SUCCESS in 7m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/df1595470b654532b5555a8739c97f60 : SUCCESS in 10m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e1dee76505c543a79285e9a9fa93bf1b : FAILURE in 2h 13m 14s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ffd938896c415b9f782333285afc86 : SUCCESS in 1h 06m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a011803b9d49cdf0859c8dd8a5c7be755eb5cb8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-05 16:49:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84123ea6587bad377b3210aa51ddaf0d3934fe8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 19:13:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce7feff72fa4a118dfdd926bc095075 : FAILURE in 6m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9d2918e734c43e591cfe27d26d46c5f : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54ce5740bca4c4faca8316db3dce644 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea1707e79494281bd3671efaf5a4849 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6b0a11c947e41a8a1ca3f79e8ba62bd : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a4ab6aa1634a5d94d61af8c95e2c97 : SUCCESS in 5m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ca333356fe14978a077ea9a03ebb115 : SUCCESS in 4m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/917801a117494befb7364f4ead2fcc31 : TIMED_OUT in 2h 15m 27s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/498811e959834d519529cb04944a913c : SUCCESS in 1h 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cb5d97d707adb4c36238525db51a47ec0f138b6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-08 23:53:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d6e18e5a65eaf2246ecb357845b26a00730394a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-08 23:54:51.000000000","message":"Abandoned\n\nfailure are note related, we are good here","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffc170f1c652a4554eeb9e9810c8c4bafea6f553","revisions":{"ffc170f1c652a4554eeb9e9810c8c4bafea6f553":{"kind":"REWORK","_number":1,"created":"2020-08-04 03:32:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/744619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/744619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/744619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/744619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/744619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/744619/1"}}},"commit":{"parents":[{"commit":"8e48d757e64d7f88058584bbf2c2898a7a94b619","subject":"Resize replicas (volume) together with primary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8e48d757e64d7f88058584bbf2c2898a7a94b619"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 02:48:38.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 03:32:54.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40225\n\nDepends-On: https://review.opendev.org/#/c/734700/\nChange-Id: Idcaab182a1b42b2506dedad4f532dae05b485cd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ffc170f1c652a4554eeb9e9810c8c4bafea6f553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ffc170f1c652a4554eeb9e9810c8c4bafea6f553"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
