)]}'
{"id":"openstack%2Ftrove~745249","triplet_id":"openstack%2Ftrove~master~Ie6a44c442e1d17587435638851a58058408d5ea8","project":"openstack/trove","branch":"master","topic":"fix-rebuild","hashtags":[],"change_id":"Ie6a44c442e1d17587435638851a58058408d5ea8","subject":"Wait for instance after rebuild","status":"MERGED","created":"2020-08-07 00:06:04.000000000","updated":"2020-08-10 00:26:02.000000000","submitted":"2020-08-10 00:24:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745249-1597019087251-62d8eb97","meta_rev_id":"af3ecc46792ce35089acb8fb8e766fca12ec7458","_number":745249,"virtual_id_number":745249,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 00:24:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-08-09 22:09:24.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-08-09 22:09:24.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 00:24:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ac0ab584d6d4a24083900677be455ba13853a1a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-07 00:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a2214024f45068da25ee612696a1095ac73cffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45ee16ab113f447c909f3779904deaef : FAILURE in 6m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6e1e5f2fc3d499988639f6d16bf71ee : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547dab04e40e4e72aa09431948fcd932 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1fef90c7f34404ab2b14c715c0e1072 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b169eee0914445afbfc1e2ae5503c0a8 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67986d7f12dc4934acc8aec5d6a7e2d5 : SUCCESS in 4m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/154eb42cf16248a0b8d2ff40b317ef12 : SUCCESS in 4m 29s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/41e6e954cd4b4baab6115d84eb5ef620 : FAILURE in 2h 03m 46s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46a43eba13d348829fdec5d533d12bbc : SUCCESS in 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e00e812dc5fe649044494b80e402572fb43e78b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-09 22:09:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa1c2202f29446306ad504c6f3ba25a495862a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-09 22:09:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"589a47e5b10f9bc9a4703a80c4b0daefa8ecf58f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 00:24:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32616e1a57b04bdb9bf64d31151f5473 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84aa98d346eb41c7804c4be2d9051160 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a5e264d788e40deb6173277ec7dc7b0 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8390242cfa94ff380ece24d3da0aa34 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e922cc9a4f43508bc9e5d69ca40b13 : SUCCESS in 7m 03s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/29cdc29880544b3aa59add3bd57480a7 : TIMED_OUT in 2h 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f56afa6b68a34e84d4110f20539c526c7d5ec0ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 00:24:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"76c3d345efc13c46d9a065ca5e6c6dabeef9b8cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 00:26:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11243c51db04920b00977525e84d0b5 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"722ac4bcc98111c153061dbcb54e824fa5a4595f","revisions":{"722ac4bcc98111c153061dbcb54e824fa5a4595f":{"kind":"REWORK","_number":1,"created":"2020-08-07 00:06:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/49/745249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/49/745249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/745249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/745249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/49/745249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/49/745249/1"}}},"commit":{"parents":[{"commit":"6b1c31e2ed513785d72b5bf73bd928453b7f5a61","subject":"Add instruction for upgrading trove guest agent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b1c31e2ed513785d72b5bf73bd928453b7f5a61"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-06 23:34:29.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-06 23:57:39.000000000","tz":720},"subject":"Wait for instance after rebuild","message":"Wait for instance after rebuild\n\nChange-Id: Ie6a44c442e1d17587435638851a58058408d5ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/722ac4bcc98111c153061dbcb54e824fa5a4595f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/722ac4bcc98111c153061dbcb54e824fa5a4595f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
