)]}'
{"id":"openstack%2Ftrove~745782","triplet_id":"openstack%2Ftrove~master~I8557782dc80b9bc691268bc96d3188bf55cdd86d","project":"openstack/trove","branch":"master","topic":"functional-test","hashtags":[],"change_id":"I8557782dc80b9bc691268bc96d3188bf55cdd86d","subject":"Fix functional test for instance access operation","status":"MERGED","created":"2020-08-12 05:36:29.000000000","updated":"2020-08-12 20:39:26.000000000","submitted":"2020-08-12 20:38:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745782-1597264700438-76a6aaaf","meta_rev_id":"8651dabd2a18db7478766efb57f63f22530f9ca3","_number":745782,"virtual_id_number":745782,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-12 20:38:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-08-12 18:20:13.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-08-12 18:20:13.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 20:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52f24d668915b87c648cd3baf6a72d84c1806a26","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-12 05:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a3863e77b0a53f7bf374fae6425ff39737bc613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 07:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/818fa5cd8a5443c9823b24c9fbedc6ce : FAILURE in 6m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4640f3a3e2e94a20b5b2e80e00a259de : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b89cecba0c34868bd9d56cada0972cc : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4bdb66a14594380a0a4ad28a6cabd87 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ab8836fccf4bbc994ac6da0cf66976 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12447669306340d5a5a677aa755064f2 : SUCCESS in 5m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9702e141d21149d8b3b99e23348c5ccc : SUCCESS in 5m 34s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c7bfbebd8b8740d3ba331d36bc6efe50 : TIMED_OUT in 2h 12m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e2297b29d8048baa874ffaf9c6a832c : FAILURE in 1h 14m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a21e766fb1b32a6c5d1ef4364b263377dc888c66","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-12 08:50:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"767888b1675fc2fe85acd13a8256f22d83519275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 11:04:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0daeea3ddc8425aaabd1a19acb3c0ed : FAILURE in 5m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2b7f7b433e44790a9b983a039c176ae : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03fc04ac151e45daa00cb054beda3629 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b21f6bd07d50420683187dcab19c02f9 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35eabcf492e94faba95bd9fe2d33d8f9 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb87cb7e56dd4706a4e2ce2173692980 : SUCCESS in 5m 37s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a547d9c9f7e44bb087867ceb8b372751 : SUCCESS in 5m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ccf8af9eeaf646a883cbad502874626a : FAILURE in 2h 07m 16s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a47612b022846eab4c0d7a4a0475aa8 : SUCCESS in 1h 07m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"297f078328424a5443904d833948ef8a635e9400","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-12 18:20:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"69a758a0249a7107eeb51677a7b79c498cd4d3ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 18:20:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e71858188ae3b210286a6c61bd3c1e37c3da4e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 20:38:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fda92e42c0d941a4b5c81b9529b6d080 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0b9a1b6205a4a8f840fcf4767ad1b5e : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24914fcc75b442e0b126db7793985efa : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2228418ef4384f038586bca6c64ccd35 : SUCCESS in 15m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de6c33d13a1f427da09958bf0deb6e6d : SUCCESS in 7m 20s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/01ca323e8f634f38ba2265172fce5b71 : FAILURE in 2h 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"febd579477cc5c47ed29bd80625305622b83627b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 20:38:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"debff5fff9b394e6f2185805116417b1de5c2101","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 20:39:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad54925bd50c485f92818003e7f2848a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4fd9cbe346de707281c3254bfbbda8f2759ec4d","revisions":{"55312515b6a011b8fd00c127e197858290d94ce0":{"kind":"REWORK","_number":1,"created":"2020-08-12 05:36:29.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/82/745782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/745782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/745782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/745782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/745782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/745782/1"}}},"commit":{"parents":[{"commit":"4de40cb5144cfd8cdc4b270f23acfdbd3eafa5be","subject":"Support to update instance access","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4de40cb5144cfd8cdc4b270f23acfdbd3eafa5be"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-11 08:56:47.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-12 01:54:54.000000000","tz":720},"subject":"Fix functional test for instance access operation","message":"Fix functional test for instance access operation\n\nChange-Id: I8557782dc80b9bc691268bc96d3188bf55cdd86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55312515b6a011b8fd00c127e197858290d94ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55312515b6a011b8fd00c127e197858290d94ce0"}]},"branch":"refs/heads/master"},"c4fd9cbe346de707281c3254bfbbda8f2759ec4d":{"kind":"REWORK","_number":2,"created":"2020-08-12 08:50:54.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/82/745782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/745782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/745782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/745782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/745782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/745782/2"}}},"commit":{"parents":[{"commit":"4de40cb5144cfd8cdc4b270f23acfdbd3eafa5be","subject":"Support to update instance access","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4de40cb5144cfd8cdc4b270f23acfdbd3eafa5be"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-11 08:56:47.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-12 06:25:31.000000000","tz":720},"subject":"Fix functional test for instance access operation","message":"Fix functional test for instance access operation\n\nChange-Id: I8557782dc80b9bc691268bc96d3188bf55cdd86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4fd9cbe346de707281c3254bfbbda8f2759ec4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c4fd9cbe346de707281c3254bfbbda8f2759ec4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
