)]}'
{"id":"openstack%2Ftrove~747568","triplet_id":"openstack%2Ftrove~master~I8041fbfdb01a7a1efa721c623ab3f43efd2cc0f0","project":"openstack/trove","branch":"master","topic":"check-user-network","hashtags":[],"change_id":"I8041fbfdb01a7a1efa721c623ab3f43efd2cc0f0","subject":"Support to check if subnet is associated with router","status":"MERGED","created":"2020-08-23 08:57:45.000000000","updated":"2020-08-24 06:11:17.000000000","submitted":"2020-08-24 06:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747568-1598249339119-f63fd487","meta_rev_id":"02b543d721cb726d9e6a35c9c91c06e350751717","_number":747568,"virtual_id_number":747568,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-24 06:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-24 03:58:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-24 03:58:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 06:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"039f1aa2cf7440e3119d45799bf4a5e39d8c81c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-23 08:57:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"849596243eda54495abf17e42ab925c206a43648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 10:53:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d25c70339924464f927d3a761509e354 : FAILURE in 6m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee1d6d9ed1674358bbb78c1ad6b8ce71 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4d355ac1724cf4830f6fa7d470d667 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae1d9bd2d948471095bd3b959feb4145 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45beeaf312884756b41112e5d4c3c82e : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0979a551fbba47a497bb25069f6ae1fe : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cb48861ff8747189b3bf62bdcd705fb : FAILURE in 5m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e8c8e4236984631bbf43ac32206f787 : SUCCESS in 6m 28s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e9dc2777a782425ab8a3c1a8dc493b5a : FAILURE in 1h 53m 30s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc28e1e5bd5048a9bfad1e362e371752 : FAILURE in 1h 02m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed029002519629cbb930769e195f6d144e23ae0a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-23 21:04:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba91f52d527b3d9a5b76f924ce02f6296a114ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-23 22:45:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c6976d88b14e6d9d98b21e703a66b9 : FAILURE in 6m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0682401a6521452a996a92aecef4b61d : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411194f6f4aa4d0b9ea615bbffc1b4aa : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f245ba550c90407ab9c37abdc214b0c8 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd7f9b96f6684f7ba325d83ecf96dbe6 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca07c1c2c794c24ad5bf0250eadca1b : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ae75e36e2aa45eb9cf275496231f517 : FAILURE in 5m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ee1186d658c43e8abd99623d9a9f161 : SUCCESS in 5m 20s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5518ab8f0be44710ad0885f3c1d7ebbd : FAILURE in 1h 39m 27s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59f28c995ba2405dbea479f05d9ed965 : FAILURE in 1h 06m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"376d06b6cce2874d527a52851aafe9e3eae096f8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-24 01:17:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcf428b83d169cbb239da029d1905cf014cbdacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 02:50:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d45360ee8cbe4c5bae1ea24bd26e4a93 : FAILURE in 5m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/855ba8bf0fe74263b74b4e7005e0d4c8 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d22133edf4c483cb57940096f48627d : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e38aa0f285264f3d8901b19f8ced5f95 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b61c4c93e97d4b8b9fec3a0d17eb0c9f : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe58525274c446c969f0c2056a65253 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/515e443b90294c02848598b651c7fa81 : SUCCESS in 6m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/caca53a98b274c368e212775a3ec44cf : SUCCESS in 6m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/692d278f7ce14774bad4c8b98c6c6175 : FAILURE in 1h 31m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d9b68d0231043b3962b26b5c6b47344 : FAILURE in 1h 09m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"beedae48a280372a4f39e6f936ccce817eeda5c4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-24 03:58:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"427c4437b9cbbb20da108cdd6f123342fa29c24c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 03:58:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28b0c55bd5235eebe1d92ea0a0271a58f2b98fd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 06:08:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac37e9e2605e4a9d9d19b531f602d81e : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f780a4add345f2b82cde334f84062d : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ebc8ffb8c8d4b068eb478c42d0dfaca : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26315e16d7344646ac9adbc337c07df1 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac32b89ef0fb48dc9aad9fd2d1afe4ba : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56f5e5b0306240cda16934a3bab34419 : SUCCESS in 5m 42s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/816d88e3bf7742dfa0f3f899c7311d6a : FAILURE in 2h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9a63e35400037e4dec163587ef2688f101f02e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 06:08:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"24e5531cafb990b7d571301325c25c408a380249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 06:11:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30734b698d64ffca50296e8129b24d1 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2081089621f4266a31535a718a8b7bd : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8daade000cc242507429a762b0df1e167801144b","revisions":{"8d9c437f7ee6cc8b311bec21eb4e31881b3aa80d":{"kind":"REWORK","_number":1,"created":"2020-08-23 08:57:45.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/747568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/747568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/747568/1"}}},"commit":{"parents":[{"commit":"111398061353c7024dc4e1c48067f3af71c5c533","subject":"Merge \"Trove upgrade guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/111398061353c7024dc4e1c48067f3af71c5c533"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-22 20:32:59.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-23 08:57:36.000000000","tz":720},"subject":"Support to check if subnet is associated with router","message":"Support to check if subnet is associated with router\n\nChange-Id: I8041fbfdb01a7a1efa721c623ab3f43efd2cc0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d9c437f7ee6cc8b311bec21eb4e31881b3aa80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8d9c437f7ee6cc8b311bec21eb4e31881b3aa80d"}]},"branch":"refs/heads/master"},"265b620959d574963e0724a7c22bd6d33b7a5752":{"kind":"REWORK","_number":2,"created":"2020-08-23 21:04:51.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/747568/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/747568/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/747568/2"}}},"commit":{"parents":[{"commit":"111398061353c7024dc4e1c48067f3af71c5c533","subject":"Merge \"Trove upgrade guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/111398061353c7024dc4e1c48067f3af71c5c533"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-22 20:32:59.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-23 20:59:42.000000000","tz":720},"subject":"Support to check if subnet is associated with router","message":"Support to check if subnet is associated with router\n\nChange-Id: I8041fbfdb01a7a1efa721c623ab3f43efd2cc0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/265b620959d574963e0724a7c22bd6d33b7a5752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/265b620959d574963e0724a7c22bd6d33b7a5752"}]},"branch":"refs/heads/master"},"8daade000cc242507429a762b0df1e167801144b":{"kind":"REWORK","_number":3,"created":"2020-08-24 01:17:59.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/68/747568/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/68/747568/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/68/747568/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/68/747568/3"}}},"commit":{"parents":[{"commit":"111398061353c7024dc4e1c48067f3af71c5c533","subject":"Merge \"Trove upgrade guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/111398061353c7024dc4e1c48067f3af71c5c533"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-22 20:32:59.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-08-24 01:17:49.000000000","tz":720},"subject":"Support to check if subnet is associated with router","message":"Support to check if subnet is associated with router\n\nChange-Id: I8041fbfdb01a7a1efa721c623ab3f43efd2cc0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8daade000cc242507429a762b0df1e167801144b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8daade000cc242507429a762b0df1e167801144b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
