)]}'
{"id":"openstack%2Ftrove~749860","triplet_id":"openstack%2Ftrove~master~I8a9fd906d6b70dfad1c38e751cadc004a5aeb13f","project":"openstack/trove","branch":"master","topic":"fix-image","hashtags":[],"change_id":"I8a9fd906d6b70dfad1c38e751cadc004a5aeb13f","subject":"Use sudo when deleting image cache dir","status":"MERGED","created":"2020-09-04 01:59:05.000000000","updated":"2020-09-04 07:17:02.000000000","submitted":"2020-09-04 07:14:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749860-1599203656553-b29784c5","meta_rev_id":"96f4aa4e5d0f7c0a87e0f88cea79440fb3389515","_number":749860,"virtual_id_number":749860,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 07:14:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-04 05:12:05.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-04 05:12:05.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 07:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e33207b2b391d91867c3c5c4a87ff3db4172e87","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-04 01:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"000e42923de28427695c8fe17c8614418215996a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 04:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/802b1b926d3e45afb0ad5d5518404c45 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fee294a0bb8d4c2d84cb6a861ee9f03a : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d28c6a5b4f64a06ab00e1649c8319df : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19656b5423f41c9885d63117d70214b : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8ab2905155499c8d98715a495a091b : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41c218ed00b45c5b00063967e002795 : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/79cd6089472a454fa752a954db159bd2 : SUCCESS in 5m 32s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7a56ac7881c34193a1dc3267e7598bdd : FAILURE in 2h 03m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6902989b8824333a1f7fb360844ad3b : FAILURE in 1h 09m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4b549cd3673d6dc03a14029f50a69c6fddc9914","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-04 05:12:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b00f3cbe21c4503d04173ab9289458cea8a8eec8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 05:12:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6949516fe4c449cb6bd8d159b3cf7f3e2d4902c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:14:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cf0bbd0335c4744aa92ac5dbb5bf8ec : SUCCESS in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9a4e3b729d4c7ca1171fb048be3f68 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd76f5ea0c94952b2bada5312c9a4cb : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0b7066774ad4ae7823601b3b48b51d1 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acaa55d850ed478c8ac2e92df11d0f7e : SUCCESS in 4m 59s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d6a9400e955a4510b7e1a90eb13e8b36 : FAILURE in 2h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e45ecb1f0bb5ce647fe73db25d3e31d4d5fa7f0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:14:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0c78a102eb9eb013ebde2e7f39cd1ab5eb011e1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 07:17:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0ba26074fe44c1a29a13b7f0936786 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec8a10e36b7a9cf725e69d75da85ba85f7818817","revisions":{"ec8a10e36b7a9cf725e69d75da85ba85f7818817":{"kind":"REWORK","_number":1,"created":"2020-09-04 01:59:05.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/60/749860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/749860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/749860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/749860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/749860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/749860/1"}}},"commit":{"parents":[{"commit":"a10d38df71c12776d4e26f1ed80d78ddb73b84a2","subject":"Merge \"Use pip3 instead of pip in trovestack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a10d38df71c12776d4e26f1ed80d78ddb73b84a2"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-04 01:58:56.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-04 01:58:56.000000000","tz":720},"subject":"Use sudo when deleting image cache dir","message":"Use sudo when deleting image cache dir\n\nChange-Id: I8a9fd906d6b70dfad1c38e751cadc004a5aeb13f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec8a10e36b7a9cf725e69d75da85ba85f7818817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ec8a10e36b7a9cf725e69d75da85ba85f7818817"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
