)]}'
{"id":"openstack%2Ftrove~749867","triplet_id":"openstack%2Ftrove~master~Ieb4d0359442cdb1dab17ffa2de74b834199ec74c","project":"openstack/trove","branch":"master","topic":"bp/six-removal","hashtags":[],"change_id":"Ieb4d0359442cdb1dab17ffa2de74b834199ec74c","subject":"Remove six.moves.urllib","status":"MERGED","created":"2020-09-04 06:00:57.000000000","updated":"2020-09-25 22:10:26.000000000","submitted":"2020-09-25 22:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749867-1601071738300-dddb1e4e","meta_rev_id":"bc231a47cb4ad3de9b78f4484369f477b7b035c1","_number":749867,"virtual_id_number":749867,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 22:08:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-25 19:53:53.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-25 19:53:53.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 19:53:53.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-09-25 22:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c7b4a13b568f1d2d26c5e5f9a21c1f5b6efeb20","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-04 06:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca05f4a329854a481dacf9f2c8c3b3eec263d651","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-04 06:15:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63ac027b7625bae3bcbb4a6c6bb6b383bd3a9019","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-04 06:27:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f844188b5140a475caefb30649b532a88509d264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 08:30:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/744f7b5a06f646f9b8686d5ac278ea69 : FAILURE in 7m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70e37c078e474fcb9c5391377461d728 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16be73a2051445192f4c01ebc9acfa3 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40ff32ad7e6e49c68b86a40aa7a1c287 : SUCCESS in 9m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b803794974247c292f5ed5f70ca0446 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205e93ef4f3241a9af3250e98be779f5 : SUCCESS in 7m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4c24da4022ad43939438c38f66e1730c : SUCCESS in 6m 32s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0a068ba579b44f2fb846c423a75ffb1d : FAILURE in 1h 56m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da243a13e8b64997a650f3017054ad6d : FAILURE in 1h 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e23f04b317f2139880ff50efb00067b2fdf8d940","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-25 12:44:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b3992f2b2b3f16da3134604f035d2f4557666237","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-25 12:58:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"99ec55620ea9f7e9fe5c9451cbca2fa785420f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:15:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5de29506ed3c470d835c710a6553cd9e : FAILURE in 8m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b5d2927df804f51babd19cd9cf5b93f : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e74f2f0fbda4702a9708948a4b1bb99 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71bba269e7fa4e029c9be55f5550774c : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9f623b6f95b4ed6a56ecacf87048849 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3a6a1dfccc4e10931b50f2b95cf26c : SUCCESS in 6m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/690785146d4b4673ac8f61e6eb7ed713 : SUCCESS in 5m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0498de0a175a49099bec5928853dbd48 : FAILURE in 2h 10m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8808a57ac1a431ea8af692e9c7d1fd8 : SUCCESS in 1h 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e112e6b263fcb8419b8dc832516b206771d1912e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 19:53:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bc372b7a9dd785d2ffb33a62f78b95e3e892b333","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 19:54:20.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c44f4daf012c40162b23ee2d6ebf6d0cf6b81c1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 22:08:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5e5e489b2398c16e1789ef290fe172f615ab4b10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 22:08:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0fba946a6824db6869ce18558980f35 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c543a8a8811a4a61836d79866b58c0e1 : SUCCESS in 10m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f41ff12fde49999bd3051517d43547 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ceb19af49e54985920069e8fa815293 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e552a41dc39481099293ced27bebd5c : SUCCESS in 6m 04s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/618533b31a6248d4a72f7891cf139a87 : FAILURE in 2h 07m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"299141e57b3a9975de07180bf0d58f58cdf274d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 22:10:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d4fc2c28d84f93bcd8913011aa6cce : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7380380fa33960e3c9a4113798a1204656db737e","revisions":{"7d19b43d3d112900e7f395588de237ef232ff200":{"kind":"REWORK","_number":1,"created":"2020-09-04 06:00:57.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/67/749867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/749867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/749867/1"}}},"commit":{"parents":[{"commit":"488d8a71e2e77457a08bab4c2870f2d110cc8b65","subject":"Merge \"Remove six.PY3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/488d8a71e2e77457a08bab4c2870f2d110cc8b65"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 05:53:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 05:53:40.000000000","tz":480},"subject":"Remove six.moves.urllib","message":"Remove six.moves.urllib\n\nReplace the six.moves.urllib with Python 3 style code\n\nChange-Id: Ieb4d0359442cdb1dab17ffa2de74b834199ec74c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d19b43d3d112900e7f395588de237ef232ff200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d19b43d3d112900e7f395588de237ef232ff200"}]},"branch":"refs/heads/master"},"a1eea1a843f75601d025be0244f7785ddcdca43c":{"kind":"REWORK","_number":2,"created":"2020-09-04 06:15:01.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/67/749867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/749867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/749867/2"}}},"commit":{"parents":[{"commit":"488d8a71e2e77457a08bab4c2870f2d110cc8b65","subject":"Merge \"Remove six.PY3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/488d8a71e2e77457a08bab4c2870f2d110cc8b65"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 05:53:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 06:13:50.000000000","tz":480},"subject":"Remove six.moves.urllib","message":"Remove six.moves.urllib\n\nReplace the six.moves.urllib with Python 3 style code\n\nChange-Id: Ieb4d0359442cdb1dab17ffa2de74b834199ec74c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1eea1a843f75601d025be0244f7785ddcdca43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a1eea1a843f75601d025be0244f7785ddcdca43c"}]},"branch":"refs/heads/master"},"756a2cb494375be71b2574505334bc8f22493815":{"kind":"REWORK","_number":3,"created":"2020-09-04 06:27:42.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/67/749867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/749867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/749867/3"}}},"commit":{"parents":[{"commit":"488d8a71e2e77457a08bab4c2870f2d110cc8b65","subject":"Merge \"Remove six.PY3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/488d8a71e2e77457a08bab4c2870f2d110cc8b65"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 05:53:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 06:26:21.000000000","tz":480},"subject":"Remove six.moves.urllib","message":"Remove six.moves.urllib\n\nReplace the six.moves.urllib with Python 3 style code\n\nChange-Id: Ieb4d0359442cdb1dab17ffa2de74b834199ec74c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/756a2cb494375be71b2574505334bc8f22493815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/756a2cb494375be71b2574505334bc8f22493815"}]},"branch":"refs/heads/master"},"5803764bb51d6b82932046c5b932f907cb6f0b69":{"kind":"REWORK","_number":4,"created":"2020-09-25 12:44:55.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/67/749867/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/749867/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/749867/4"}}},"commit":{"parents":[{"commit":"488d8a71e2e77457a08bab4c2870f2d110cc8b65","subject":"Merge \"Remove six.PY3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/488d8a71e2e77457a08bab4c2870f2d110cc8b65"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 05:53:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-25 12:40:39.000000000","tz":480},"subject":"Remove six.moves.urllib","message":"Remove six.moves.urllib\n\nReplace the six.moves.urllib with Python 3 style code\n\nChange-Id: Ieb4d0359442cdb1dab17ffa2de74b834199ec74c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5803764bb51d6b82932046c5b932f907cb6f0b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5803764bb51d6b82932046c5b932f907cb6f0b69"}]},"branch":"refs/heads/master"},"7380380fa33960e3c9a4113798a1204656db737e":{"kind":"REWORK","_number":5,"created":"2020-09-25 12:58:22.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/67/749867/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/67/749867/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/67/749867/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/67/749867/5"}}},"commit":{"parents":[{"commit":"488d8a71e2e77457a08bab4c2870f2d110cc8b65","subject":"Merge \"Remove six.PY3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/488d8a71e2e77457a08bab4c2870f2d110cc8b65"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-09-04 05:53:40.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-09-25 12:55:47.000000000","tz":480},"subject":"Remove six.moves.urllib","message":"Remove six.moves.urllib\n\nReplace the six.moves.urllib with Python 3 style code\n\nChange-Id: Ieb4d0359442cdb1dab17ffa2de74b834199ec74c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7380380fa33960e3c9a4113798a1204656db737e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7380380fa33960e3c9a4113798a1204656db737e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
