)]}'
{"id":"openstack%2Ftrove~750138","triplet_id":"openstack%2Ftrove~master~Icf08b7dc82ce501d82b45cf5412256a43716b6ae","project":"openstack/trove","branch":"master","topic":"postgresql","hashtags":[],"change_id":"Icf08b7dc82ce501d82b45cf5412256a43716b6ae","subject":"Postgresql: Backup and restore","status":"MERGED","created":"2020-09-07 08:41:12.000000000","updated":"2020-09-07 21:01:17.000000000","submitted":"2020-09-07 21:00:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750138-1599512410246-9d651cf1","meta_rev_id":"af69e8977bd8c43a3c0aaf20cc52f59e28f5fb75","_number":750138,"virtual_id_number":750138,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-07 21:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-07 18:59:13.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-07 18:59:13.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-07 21:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b11c6d4e205e4b5f19b25f5057d7d0c6b001d1a3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-07 08:41:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e48315733b1a8b06991742b37492487dbe949c14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:49:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac10cac6c9094be79dc845635e191f0e : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f04a2971d0b241cc812f1b379fc5d689 : FAILURE in 6m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a69644ac54b645249f6c500453d9d06e : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8c444f0c0e443fa2ec4e7328c1d675 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83c6585e764a448d8f5213495a0aa1b2 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e5720ac0139410aa5f9ff5e478f1441 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a0776c138b47ccbb172a745e8884ca : SUCCESS in 5m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8c0bea16561d4ef1a56c03e51ef0ca39 : SUCCESS in 6m 57s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/30e0f22d0bbc4c34b4f64af2169c795c : FAILURE in 2h 02m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a4fea5d74944cfbb7eaa32001738236 : FAILURE in 1h 08m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6aea189d00f9af5f982541fbab199d167e33477","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-07 18:59:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"977edc6eb2310fa9dbb4b21e649ae422b2520e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 18:59:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"156d498cfd163302d0674e340ad71490ee09fc79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 21:00:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4f02b6f3aad4662ac1a2f0b5e8ac45c : SUCCESS in 3m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1be10dc9aa2f4fd1a5fd8595bb854aec : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7042976b6daf4df2adba8f2d554b80dc : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fd22ae929b24a92b8bf733963d7f76b : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a489146865945888408f6d43bf197bf : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4135bcc53ae5480d89a9e08e7ffec05f : SUCCESS in 4m 48s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8c23e5c79b464c29a88c5e91f86c1a9d : FAILURE in 1h 54m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d1a4ab8a894f236e78dc93627519edf7ea2f7d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 21:00:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"152d185f8add825299c00d9ac1bb5fa2177d2101","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 21:01:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee0bc7580f241998d36afb3aa952c50 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fb41b5198c865b46a02dd72501d12e60ec10dd6","revisions":{"4fb41b5198c865b46a02dd72501d12e60ec10dd6":{"kind":"REWORK","_number":1,"created":"2020-09-07 08:41:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/38/750138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/38/750138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/750138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/750138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/38/750138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/38/750138/1"}}},"commit":{"parents":[{"commit":"768ec34dfef660f133f87218a6246a9ce111bcb5","subject":"Merge \"Use sudo when deleting image cache dir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/768ec34dfef660f133f87218a6246a9ce111bcb5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-01 22:10:23.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-07 08:40:56.000000000","tz":720},"subject":"Postgresql: Backup and restore","message":"Postgresql: Backup and restore\n\nChange-Id: Icf08b7dc82ce501d82b45cf5412256a43716b6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fb41b5198c865b46a02dd72501d12e60ec10dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4fb41b5198c865b46a02dd72501d12e60ec10dd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
