)]}'
{"id":"openstack%2Ftrove~750940","triplet_id":"openstack%2Ftrove~master~Ia00032074dc44a6fbfc1e2d5ab16d1734a1a732c","project":"openstack/trove","branch":"master","topic":"postgresql-replication","hashtags":[],"change_id":"Ia00032074dc44a6fbfc1e2d5ab16d1734a1a732c","subject":"[Postgresql] Create replica","status":"MERGED","created":"2020-09-10 09:55:23.000000000","updated":"2020-09-11 23:29:10.000000000","submitted":"2020-09-11 23:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750940-1599866875171-c09e01d6","meta_rev_id":"7f4a63d4e7d5daf7a0fee8db6368292325e7f070","_number":750940,"virtual_id_number":750940,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 23:27:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-11 21:12:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-11 21:12:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 23:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4817fbf9e8080e1a2df81f27946413596e54d856","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-10 09:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43594c3e78c1644890dd05d68ce0e383e954e7cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 11:45:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1268ee7bdd0b44dc8d548cff403199cc : FAILURE in 9m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6169eb4253514d87972b6ce44616d523 : FAILURE in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7b2febf55e4213b5113c3ccd920d5e : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00ddc5f6304e4a5eaa67d3b7c0f69480 : FAILURE in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/772a0bd0c745429392d20f582b39c88f : FAILURE in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2a588740d24878b43ec334ebe8f7db : SUCCESS in 7m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff8af990090043a0b794e19313c479df : SUCCESS in 7m 00s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b9e55693e4a343a1a6b36d573c8db90a : FAILURE in 1h 45m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c9d2691f0594265b02f855329fa22d1 : SUCCESS in 1h 06m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32d1709bed402f54ff555c44b375f828f6e2ede2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-10 22:30:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd096d33f5fc8c4b06e995cd70936ad26de8eef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 00:47:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/544d70bd52504155adb6d7a21d95be52 : FAILURE in 6m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53ac04dd5669483fbbebc1b43c54cbcc : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6364691fcd0b411eba1b6b197d7584ce : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/253b0dcd91ca49398d3c72a36b4c7926 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce4c102985b5444088a674a204914ca5 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eac0e8789def4989849d9fac1a3f6ccd : SUCCESS in 6m 20s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da839cef47ed4e81b5a9cd6a58b09da4 : SUCCESS in 6m 04s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f68dd38d64604bcca6726be5ce54865e : FAILURE in 2h 11m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4950b4b0022b43da873732d50a12667a : SUCCESS in 1h 04m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8505ef577a44d05da82df621e8b7ebabe2c58fd","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-11 04:40:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1144820321dcc322c34eed680033488c8b87e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 06:42:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05476bfd594444588ce9cd6d037a66e1 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbb875b37b9f4909b81568f752ef753c : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3750912b6f94245a7a6bf866bb7ce3b : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e3d4832d0534da38c45cc655a6ff732 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2ecbd9e56ce41ba95608282ab66034a : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8067fd682d74cb38391985da588a16f : SUCCESS in 5m 43s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a884fbb608d24de88b881f7defa174a4 : SUCCESS in 5m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c2df0c308ae34dfc96eabdc0325e8413 : FAILURE in 2h 00m 16s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3eba40aba39478da85eb536f5c17440 : SUCCESS in 1h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"955cb046630e74964d7a99c7195163e3c58fbe04","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-11 21:12:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"447f1db25b50f29b78af45ebcc4d2257b8c3e9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:22:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f238879646281545d19d9bcbfee1caf550a6085d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:27:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9be02ad3e0744f3ab14763679bf5654f : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09fb48820bd24239b7305baf3d9d6ee4 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df2db9578b2b4ce8828f1b2b9d37f442 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86eace79d8584d2bb84823330881a1d7 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0071e8c03764b9c8b0dd2590edbd064 : SUCCESS in 6m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/73a9212e64834f09a6a58afe9985ae10 : FAILURE in 2h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3fe2d83b9087bf8d1a39110fcc4d3eed518ad689","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:27:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fb292aa876f08696d81a59582d14c10ab5e8a2b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 23:29:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fb0b6b81eb4013b1bb2f75d050aefd : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5482c54645fcea2af6b61837e8ff78cd77e1810c","revisions":{"8687bf2d0dae59ce7cde50402787a1904521f426":{"kind":"REWORK","_number":1,"created":"2020-09-10 09:55:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/40/750940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/750940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/750940/1"}}},"commit":{"parents":[{"commit":"8761f327fe8e4fca55d8fd667f7b9cf279d29299","subject":"Merge \"[API doc] Add replica_of param for getting instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8761f327fe8e4fca55d8fd667f7b9cf279d29299"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-08 11:03:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-10 04:13:06.000000000","tz":720},"subject":"[Postgresql] Create replica","message":"[Postgresql] Create replica\n\nChange-Id: Ia00032074dc44a6fbfc1e2d5ab16d1734a1a732c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8687bf2d0dae59ce7cde50402787a1904521f426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8687bf2d0dae59ce7cde50402787a1904521f426"}]},"branch":"refs/heads/master"},"27ca121fd03b88da6a7bf6133597b81ac6be9f48":{"kind":"REWORK","_number":2,"created":"2020-09-10 22:30:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/40/750940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/750940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/750940/2"}}},"commit":{"parents":[{"commit":"8761f327fe8e4fca55d8fd667f7b9cf279d29299","subject":"Merge \"[API doc] Add replica_of param for getting instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8761f327fe8e4fca55d8fd667f7b9cf279d29299"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-08 11:03:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-10 22:29:42.000000000","tz":720},"subject":"[Postgresql] Create replica","message":"[Postgresql] Create replica\n\nChange-Id: Ia00032074dc44a6fbfc1e2d5ab16d1734a1a732c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27ca121fd03b88da6a7bf6133597b81ac6be9f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/27ca121fd03b88da6a7bf6133597b81ac6be9f48"}]},"branch":"refs/heads/master"},"5482c54645fcea2af6b61837e8ff78cd77e1810c":{"kind":"REWORK","_number":3,"created":"2020-09-11 04:40:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/40/750940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/750940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/750940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/750940/3"}}},"commit":{"parents":[{"commit":"8761f327fe8e4fca55d8fd667f7b9cf279d29299","subject":"Merge \"[API doc] Add replica_of param for getting instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8761f327fe8e4fca55d8fd667f7b9cf279d29299"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-08 11:03:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-11 01:48:32.000000000","tz":720},"subject":"[Postgresql] Create replica","message":"[Postgresql] Create replica\n\nChange-Id: Ia00032074dc44a6fbfc1e2d5ab16d1734a1a732c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5482c54645fcea2af6b61837e8ff78cd77e1810c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5482c54645fcea2af6b61837e8ff78cd77e1810c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
