)]}'
{"id":"openstack%2Ftrove~751029","triplet_id":"openstack%2Ftrove~master~I51fab3f716dbd77da21c9cbbf91ec73c673c5c8b","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I51fab3f716dbd77da21c9cbbf91ec73c673c5c8b","subject":"Fix l-c testing for Ubuntu Focal (py38)","status":"ABANDONED","created":"2020-09-10 15:35:06.000000000","updated":"2026-09-17 17:27:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b2c1b4166f8146f3a2c7be98999b48c0fd0e304","_number":751029,"virtual_id_number":751029,"owner":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-10 16:53:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 16:53:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b44ed534a46989cc7ecc233b670d0e9dc13f983","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-09-10 15:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6d2726b77aa723d15d08d12b693b7b6ec8484b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 16:53:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f999120b98ff415ebdde8aea78416e71 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82aab7af66b34bf59ccb7a83258466ac : FAILURE in 7m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4ed5a1c19654b2a814a37a5fd87141c : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5b23a097e64eaa9192e320f0dcb033 : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a071238a901d472f879446bf3a766543 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b9d52f089f4b77a9fd4d6449acb4b6 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7047b985518442ee983227e9f541da22 : SUCCESS in 5m 47s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/63180045b7ae4b5cbaffbe5a4f7e021b : FAILURE in 1h 11m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01e0d513845342159c9a17d29680eefe : FAILURE in 1h 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b2c1b4166f8146f3a2c7be98999b48c0fd0e304","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:27:06.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ac460374752427ee01551ca07890e1cbd3c26b6","revisions":{"6ac460374752427ee01551ca07890e1cbd3c26b6":{"kind":"REWORK","_number":1,"created":"2020-09-10 15:35:06.000000000","uploader":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"ref":"refs/changes/29/751029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/751029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/751029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/751029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/751029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/751029/1"}}},"commit":{"parents":[{"commit":"8761f327fe8e4fca55d8fd667f7b9cf279d29299","subject":"Merge \"[API doc] Add replica_of param for getting instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8761f327fe8e4fca55d8fd667f7b9cf279d29299"}]}],"author":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-10 15:34:48.000000000","tz":480},"committer":{"name":"zhoulinhui","email":"df.some@foxmail.com","date":"2020-09-10 15:35:04.000000000","tz":480},"subject":"Fix l-c testing for Ubuntu Focal (py38)","message":"Fix l-c testing for Ubuntu Focal (py38)\n\nChange-Id: I51fab3f716dbd77da21c9cbbf91ec73c673c5c8b\nStory: 2007865\nTask: 40199\nCloses-Bug: #1886298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ac460374752427ee01551ca07890e1cbd3c26b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ac460374752427ee01551ca07890e1cbd3c26b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
