)]}'
{"id":"openstack%2Ftrove~751586","triplet_id":"openstack%2Ftrove~master~Ia42c72dd3ce88b083b55fa73936b81c78da21166","project":"openstack/trove","branch":"master","topic":"fix","hashtags":[],"change_id":"Ia42c72dd3ce88b083b55fa73936b81c78da21166","subject":"Fix some API calls for PostgreSQL","status":"MERGED","created":"2020-09-12 23:45:08.000000000","updated":"2020-09-13 09:18:59.000000000","submitted":"2020-09-13 09:17:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751586-1599988665749-e889b14a","meta_rev_id":"57b993d0e8c0c7049d89af98202996a19e99bbe4","_number":751586,"virtual_id_number":751586,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-13 09:17:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-13 07:16:46.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-13 07:16:46.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-13 09:17:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55aa23709c8d903f48112dc7c32fdbe5565933bb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-12 23:45:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1999d82d1b4c0466bfcfb52877c37c844b04b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 01:39:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7770a1a1bbb4c1ea16f005514495926 : FAILURE in 5m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f85f885ae2945d5817fbf1dacbe040e : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7b6e48fdbe14ee8a4bcde45d18ec5b9 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3267ff007c464452b3e6c56933b4e15d : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73bc3e2d5dbb4201aa949699e11db5cf : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40ed66c341048e984566724597a5ecf : SUCCESS in 6m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8691295673424eb5878305e175c93a5a : SUCCESS in 5m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5b05056ce311427a800bcc4d64806679 : FAILURE in 1h 46m 32s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e110dd79214cd18baa11eaf7b5b20b : SUCCESS in 1h 02m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c992f6eeb96ea6fca56abbc517d740fa387e9a4d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-13 07:16:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c4eb89617724d403e163b3fa4532b39a8c44d37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 07:17:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20210b96eeeb3f2a94b068feb294a793b05babb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 09:17:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/906e1219a4a04fc5a3de078ba01ef938 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5fef46bdeb410899c595bd4404146f : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87d13d60b9954d319a5fe0288a5d7306 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3954567cafc4994af6e9f93d20759c3 : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02675f699684c59a7e36991678c2880 : SUCCESS in 5m 58s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/90da3910687541cf9263a8a6892e9475 : FAILURE in 1h 59m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"885eaa4b0c2e045b7586b6e612c9281e999f2d59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 09:17:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1543b2d4aa4aab2649125f2862ad3c49f7864f24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-13 09:18:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b623d7ecd1c40cfacd00740de26006f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df4f74769ee731c57cf8dcef01dec8946baa58c6","revisions":{"df4f74769ee731c57cf8dcef01dec8946baa58c6":{"kind":"REWORK","_number":1,"created":"2020-09-12 23:45:08.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/86/751586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/86/751586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/751586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/751586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/86/751586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/86/751586/1"}}},"commit":{"parents":[{"commit":"5482c54645fcea2af6b61837e8ff78cd77e1810c","subject":"[Postgresql] Create replica","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5482c54645fcea2af6b61837e8ff78cd77e1810c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-12 05:06:23.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-12 23:44:54.000000000","tz":720},"subject":"Fix some API calls for PostgreSQL","message":"Fix some API calls for PostgreSQL\n\nChange-Id: Ia42c72dd3ce88b083b55fa73936b81c78da21166\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/df4f74769ee731c57cf8dcef01dec8946baa58c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/df4f74769ee731c57cf8dcef01dec8946baa58c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
