)]}'
{"id":"openstack%2Ftrove~751911","triplet_id":"openstack%2Ftrove~master~I99fb3d0c7ea7036bbb622d86eef9f31e53065a0a","project":"openstack/trove","branch":"master","topic":"backup-project","hashtags":[],"change_id":"I99fb3d0c7ea7036bbb622d86eef9f31e53065a0a","subject":"Add project_id to backup response","status":"MERGED","created":"2020-09-14 22:09:13.000000000","updated":"2020-09-16 01:45:57.000000000","submitted":"2020-09-16 01:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751911-1600220670691-8dffc5fa","meta_rev_id":"7213f41da858abc52aa98cfa2c712ffead1ac711","_number":751911,"virtual_id_number":751911,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 01:44:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-15 21:47:53.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-15 21:47:53.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 01:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64450a3dd492257dea68d72b64f7d84c9f6afd31","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-14 22:09:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12eca7793ccdced0d1ca5d167b199b06c0655685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 00:24:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ca95d5c1c3de49adb1079c0519af69f1 : SUCCESS in 11m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f38f90f97a0e48b09cf5e152faf56b8c : FAILURE in 5m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b83e82dc84f4a35bc6f015f9905d627 : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa31092fe924041b5e58ad4168386b4 : FAILURE in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc8f2c23c9fe491f9763753acf03d630 : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526d93a0bcc34d17be45f42d7dd39983 : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e51ca1e8fa441d1b819cebca94fd469 : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5931e58620a249319701c0da47d5465e : FAILURE in 4m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f05e5b29890d4123b2868e83b23c4e33 : TIMED_OUT in 2h 13m 33s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd4748ed79b44cf18628991a9d3d2f46 : FAILURE in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3d65568cd462ac6c3d4bfb2855543a07b841b91","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-15 04:46:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5cd8ee94d0f69b1d5aea8c8a2c52cea28cf16efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 05:05:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/44832f5c9ce246a287eb4cd8ecb666db : FAILURE in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf27ead744a24e6bb6fa6ece8a1fa5b8 : FAILURE in 3m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bb30abf660b453ea330f59429c31717 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833d9d40ed63443aae5386bfa701456c : FAILURE in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d9fd57a4198458fb2a99f90bcc687af : FAILURE in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af550573ba254cc49d2ac03c60e947bc : FAILURE in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbc4c750f99c414ca2bd1c9d0b634fe4 : FAILURE in 4m 09s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cac9c850e02944fd8341b9fe9b1b2a57 : FAILURE in 3m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d7ca336352f8432b90d228074e6964eb : FAILURE in 12m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36affb9a452c4c0b929b59fac586f9da : FAILURE in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"505342137cab494ecbf4efa96d6850e0e69f364c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-15 21:47:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"627b717c6e43fa33ff94dc98df582743c5060028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 23:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93d74a5a0f9a41ccb18ad33ad8ffa109 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85ee6dcac29f4fdd916fd04a14602791 : FAILURE in 6m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0543c483a65f4345a400dcdc36c4686c : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dddf2c433374ea39dc2a4fffc27df84 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad67421fe868475bbcdded883d227844 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/539136f5ad2f46e2a2194f42a7a4b99c : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac221731bbc44a2cb93cd450e75d4331 : SUCCESS in 6m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ea63710ba52477e8dee7f0668d68b47 : SUCCESS in 5m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2e8aea105891472da579c1a8ec958612 : FAILURE in 1h 35m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15d0c0104fe141c8b84ddc838749eea2 : FAILURE in 1h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db4e8e36e56381c0c78f0e66a81f4c3cd8e6e636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 23:26:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ea704891f8fcd0514faef016d385828abf4fef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 01:44:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c1c37e65f81fb8f53eee8ee6ee10a79eff3c32f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 01:44:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6cc67ad5597548c19d9bb0ba336a9bcb : SUCCESS in 5m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bf0d77a90d44ac8b4642dcd72cca21e : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd5b7aca6aa424c91dc20f3df3b19a3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/289f1ff6f1114fcf893757408eb20d33 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fc253b436894a81ab1b280b0bc77318 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74c875fd9274db489ffe629dca6fc77 : SUCCESS in 5m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/65026ff04907480cb4b6c1bf382400af : TIMED_OUT in 2h 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8778612ce96b1671d40c1cae3fffa51e62df2fa3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 01:45:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/432095e49b0b4f94a9357d9a5c8d889b : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f0086824fa4f679ce76a413f07e078 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86a0102a003df890a8d693a60b1e55ba7b0328f1","revisions":{"86a0102a003df890a8d693a60b1e55ba7b0328f1":{"kind":"REWORK","_number":1,"created":"2020-09-14 22:09:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/11/751911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/11/751911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/751911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/751911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/11/751911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/11/751911/1"}}},"commit":{"parents":[{"commit":"2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0","subject":"Merge \"Support getting backups of a specific project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-14 22:03:53.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-14 22:08:57.000000000","tz":720},"subject":"Add project_id to backup response","message":"Add project_id to backup response\n\nChange-Id: I99fb3d0c7ea7036bbb622d86eef9f31e53065a0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86a0102a003df890a8d693a60b1e55ba7b0328f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86a0102a003df890a8d693a60b1e55ba7b0328f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
