)]}'
{"id":"openstack%2Ftrove~752030","triplet_id":"openstack%2Ftrove~master~I69f111f6caf14ad5adbefacd8328b45d7cba77b3","project":"openstack/trove","branch":"master","topic":"json-type-in-db","hashtags":[],"change_id":"I69f111f6caf14ad5adbefacd8328b45d7cba77b3","subject":"Change access column type for instance","status":"MERGED","created":"2020-09-15 12:07:02.000000000","updated":"2020-09-16 15:15:35.000000000","submitted":"2020-09-16 15:13:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752030-1600269197554-079f7ff7","meta_rev_id":"6c383eda6b3cf1b5fdbfc8d21c10170910d92daa","_number":752030,"virtual_id_number":752030,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 15:13:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-16 13:22:15.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-16 10:25:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-16 13:22:15.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 10:25:43.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"state":"REVIEWER"},{"updated":"2020-09-16 15:13:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"375d3fd72edc75fcf82be57fe7f3be5cffe21dc8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-15 12:07:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7660d04ae1e97a853651640075bf2968208c6628","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-15 12:17:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b00c6fd396d732f39339f0206a4e33687ab208c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:01:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2803a64703c48d1a755502982975c58 : FAILURE in 10m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3196ffe3be384f63b2ce68ac597a06a3 : POST_FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928f3d1406ce40019d18f0d83d16c6c2 : FAILURE in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adbadcd85ef7410bbfc0c1beae3d0796 : FAILURE in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d99dece9a1b94fe4955b0afad059b8a3 : POST_FAILURE in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0495b152a244b78fe73f165576f07a : SUCCESS in 5m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9027b3b0a807454db66ede6a67e451f1 : FAILURE in 3m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b74d719ce8db4113beb3d16a41a66490 : FAILURE in 16m 12s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e797dbf3aa6342b9a03c9d77376207de : FAILURE in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e20e24d18e34ff37504307990b13ef04ea4a4c0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-15 15:30:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"900b855a2964594babefd8220e877680d1dfe80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 15:54:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e0906ef8234637b34b45b67ddf8ec4 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7b097ec0dca41ddb42de257e0ad8a90 : POST_FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e8e3a444584df4b957b2e94f9a201f : FAILURE in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93a7cb5597cb4817878460f46ec5b740 : FAILURE in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90cfa0bdb0354ba8b08122fa772d653c : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a32c7402ebd4ee5b6c491188987b914 : FAILURE in 4m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c52601f6cafe41a4aab1fea0594f3ce8 : FAILURE in 4m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/85b053ab7d024bc19b44033c27948178 : FAILURE in 16m 56s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffc2d2e9ef1544cbad4aa847fc5dcccf : FAILURE in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3872e23f8d7b21e6921d0393afe64ec51e5bfc4b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-15 21:48:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb87334fc91870230f41ec701498c2fb84449042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 22:35:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5cdb9c225cb47439b40890c08191d3a : FAILURE in 6m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cce1e23e50f4600aaf5bf1bdd37514b : FAILURE in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e734d0286346fb8751454ae483cb52 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f027e7e1164f68a7273abd1fc74772 : FAILURE in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb6d7d80f2cd4f288b2e8e0cff49bd42 : FAILURE in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb81a85c2ca422da7a07fac9cc66518 : SUCCESS in 6m 02s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/36e5b3cfa4fe4f0aba64c470207a8bec : SUCCESS in 6m 08s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/152c7de1c85948a6a349f13d05e3e7d5 : FAILURE in 44m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e26ac8c67a654118ba8ed673ee67779f : FAILURE in 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef1b23a44d23dbac9c7d9c1adb41b1397c711099","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-09-16 04:10:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15f653e1bdc3ae9c16d624fb033eb35a483f8817","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-16 10:16:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"43fe9891f768018c4feeb1925f4e0f739eace075","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-09-16 10:25:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9f4b9ca69e2aec9e5553e5f80e82ed6c726da5e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:59:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52f8d228ae0a46e9b3312e935474190e : FAILURE in 6m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0539ce5cf3054f56a62f32d7b1594cf1 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fcbc768ce742578ac577f31a537de3 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a750617fc1a444b8989cf27b1d456aa : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e6ddba79bf8416aa71f11826b38d685 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625c4bdd20f348e892dac9a78d7bee20 : SUCCESS in 5m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fd3dcd081d6949f9b3c98ef12d629818 : SUCCESS in 5m 47s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4aff36106f6442bb885fa3bcdab477e8 : FAILURE in 1h 41m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/736c8b8ae95e446294bab94ea0232d91 : SUCCESS in 1h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"852888b6766bc45cfc953b891bfd46941c7971cc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-16 13:22:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f70a90070e7c870cc2d4d0475310a15ea4804f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:24:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d674a9f7a6bb8fdce0c1f86a3ae81e65cbb38d2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 15:13:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6458b018e202c5eb55ce9c7ff719b1a2a6d88add","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 15:13:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32bfdd96f1e142b788321c400f7add0c : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd34c2c43544dcdb9c5958a67a73c9a : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77ce3f3298d64a7c8cbe629cf9be42e1 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af7fd0e9858f4f5fa31fb8bd8a83a258 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9c66fd88e84464aadac0c41e1d4920 : SUCCESS in 6m 32s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8d5c41d615704e2ab3663060a33d82f7 : FAILURE in 1h 46m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d369786478a6ea0401f5732ca9e4c373c048ac1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 15:15:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4460c4271c04116b0187d6b9dc6f875 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3aa6847296a57d6c79aac61dca15d1387baf3594","revisions":{"ef39abbdad41f37e3bfbb6c10ce5159d5383cb9a":{"kind":"REWORK","_number":1,"created":"2020-09-15 12:07:02.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/752030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/752030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/752030/1"}}},"commit":{"parents":[{"commit":"2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0","subject":"Merge \"Support getting backups of a specific project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 10:13:43.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 12:00:49.000000000","tz":720},"subject":"Change access column type for instance","message":"Change access column type for instance\n\nThe sqlalchemy.types.JSON is not compatible between different database\nengines, use TEXT type instead.\n\nStory: #2008145\nTask: #40886\n\nChange-Id: I69f111f6caf14ad5adbefacd8328b45d7cba77b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef39abbdad41f37e3bfbb6c10ce5159d5383cb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ef39abbdad41f37e3bfbb6c10ce5159d5383cb9a"}]},"branch":"refs/heads/master"},"7cdd86174c18843a58fe75c39df87e7a2551361f":{"kind":"REWORK","_number":2,"created":"2020-09-15 12:17:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/752030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/752030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/752030/2"}}},"commit":{"parents":[{"commit":"2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0","subject":"Merge \"Support getting backups of a specific project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 10:13:43.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 12:16:24.000000000","tz":720},"subject":"Change access column type for instance","message":"Change access column type for instance\n\nThe sqlalchemy.types.JSON is not compatible between different database\nengines, use TEXT type instead.\n\nRef:\n* https://docs.sqlalchemy.org/en/13/core/type_basics.html#sqlalchemy.types.JSON\n* https://mariadb.com/kb/en/json-data-type/\n\nStory: #2008145\nTask: #40886\n\nChange-Id: I69f111f6caf14ad5adbefacd8328b45d7cba77b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7cdd86174c18843a58fe75c39df87e7a2551361f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7cdd86174c18843a58fe75c39df87e7a2551361f"}]},"branch":"refs/heads/master"},"aefeb5941a092c3833da9bbcc1a3737ec30532a2":{"kind":"REWORK","_number":3,"created":"2020-09-15 15:30:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/752030/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/752030/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/752030/3"}}},"commit":{"parents":[{"commit":"2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0","subject":"Merge \"Support getting backups of a specific project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 10:13:43.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 15:30:35.000000000","tz":720},"subject":"Change access column type for instance","message":"Change access column type for instance\n\nThe sqlalchemy.types.JSON is not compatible between different database\nengines, use TEXT type instead.\n\nRef:\n* https://docs.sqlalchemy.org/en/13/core/type_basics.html#sqlalchemy.types.JSON\n* https://mariadb.com/kb/en/json-data-type/\n\nStory: #2008145\nTask: #40886\n\nChange-Id: I69f111f6caf14ad5adbefacd8328b45d7cba77b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aefeb5941a092c3833da9bbcc1a3737ec30532a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aefeb5941a092c3833da9bbcc1a3737ec30532a2"}]},"branch":"refs/heads/master"},"3aa6847296a57d6c79aac61dca15d1387baf3594":{"kind":"REWORK","_number":4,"created":"2020-09-16 10:16:08.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/752030/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/752030/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/752030/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/752030/4"}}},"commit":{"parents":[{"commit":"2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0","subject":"Merge \"Support getting backups of a specific project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2cd5bb6f831f7bee44d698ecacaa7006b12f0ec0"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-15 10:13:43.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-16 10:15:58.000000000","tz":720},"subject":"Change access column type for instance","message":"Change access column type for instance\n\nThe sqlalchemy.types.JSON is not compatible between different database\nengines, use TEXT type instead.\n\nRef:\n* https://docs.sqlalchemy.org/en/13/core/type_basics.html#sqlalchemy.types.JSON\n* https://mariadb.com/kb/en/json-data-type/\n\nStory: #2008145\nTask: #40886\n\nChange-Id: I69f111f6caf14ad5adbefacd8328b45d7cba77b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3aa6847296a57d6c79aac61dca15d1387baf3594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3aa6847296a57d6c79aac61dca15d1387baf3594"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
