)]}'
{"id":"openstack%2Ftrove~752347","triplet_id":"openstack%2Ftrove~stable%2Fussuri~I451551b7cb4e812d81b1341c5b82da3abf7b8d86","project":"openstack/trove","branch":"stable/ussuri","topic":"ussuri-fix-swift-404","hashtags":[],"change_id":"I451551b7cb4e812d81b1341c5b82da3abf7b8d86","subject":"[Backup] Ignore 404 error when deleting swift objects","status":"MERGED","created":"2020-09-16 23:36:36.000000000","updated":"2020-09-17 07:33:11.000000000","submitted":"2020-09-17 07:31:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752347-1600327879587-c6eda660","meta_rev_id":"f1f0aa093ba0eb384f81d2fa65ef0bbf277e48af","_number":752347,"virtual_id_number":752347,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 07:31:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-17 05:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-17 05:13:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 07:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b188e12e0fcbf3fd9446be20dd3131e86cfc59a1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-16 23:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47d2b1a7ab98826ce3c97876a03c37d23778ab10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 01:52:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ac33566ad8f4d9d980455d7e2ce1143 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/985d7d2975c7446bbc8a39f596ae7e52 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5de4ddbe7e64628ab51fd1f39021a29 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffac1e8647234f058cca8ba0ee3b937f : SUCCESS in 11m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13b0190d19e54941a8149d7d593c2cb4 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46f85dcade04265b9cd3687cfd8c497 : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/25a382d8b57e47dd9ff94bf6da1f7cb4 : SUCCESS in 5m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9f3a6c334bac472ba7670d8faef45d71 : SUCCESS in 37m 12s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c24eebfaa42c424abfa41b14af47a670 : SUCCESS in 2h 05m 26s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/b310203f0a2f4eda945b12fdf5ce98c9 : SUCCESS in 2h 11m 45s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/7b8303d77e5f4af9831480df1ec94d5d : SUCCESS in 1h 26m 55s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e2d84a0d856945ae933d6fd0aba0e437 : SUCCESS in 1h 39m 01s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/e906e29f838b4cd9bb0e353f8c627c8a : FAILURE in 41m 39s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/252172869f1841b9925990d2d8bd9898 : FAILURE in 33m 48s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/802f4701080a4294b8a5fda71af132ad : FAILURE in 46m 57s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0240e0b01ef4e90b4f4341a92496e1b : FAILURE in 45m 47s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/fe40c70903a14171a7075bee25d7cea5 : SUCCESS in 1h 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ecf67cb8440895f80820837628afc88adad8e6f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-17 05:13:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67a478b82d0db9bdb17549aa842605ae811862cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 05:13:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78e7e9a98417a54e49e9627833225a703c21c9ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:31:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6af189ea3bfc4491a9a4882bc926e13b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1199722f48cf43e683606f6759e46176 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45995c9ebd6a41cbbd7015ef955dde40 : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e9c3c7cfe4443579c0d1165b678a5a7 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39854c1710d341aba19bc23256470594 : SUCCESS in 5m 44s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6eb0a82a8acc443bbaff56069827613b : SUCCESS in 1h 55m 53s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/dc39635b072446ecabfdd24839059c96 : SUCCESS in 2h 16m 03s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/d6ad42f81fe242ec9cd327afd84313b1 : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2c36b353751e92feeca5ea9f000780da21535788","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:31:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"268968f70fced7ddea54e9ec0438c70129143ec6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:33:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e92c58a5724632a443b5050e127186 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f31ac7463a5ee93d139eecd62c41132bbd0ce520","revisions":{"f31ac7463a5ee93d139eecd62c41132bbd0ce520":{"kind":"REWORK","_number":1,"created":"2020-09-16 23:36:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/47/752347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/752347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/752347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/752347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/752347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/752347/1"}}},"commit":{"parents":[{"commit":"695a63e228548349cfc8d7ced75ecc53e1b1f984","subject":"Do not operate floating IP for relication cluster","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/695a63e228548349cfc8d7ced75ecc53e1b1f984"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-16 23:36:28.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-16 23:36:28.000000000","tz":720},"subject":"[Backup] Ignore 404 error when deleting swift objects","message":"[Backup] Ignore 404 error when deleting swift objects\n\nChange-Id: I451551b7cb4e812d81b1341c5b82da3abf7b8d86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f31ac7463a5ee93d139eecd62c41132bbd0ce520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f31ac7463a5ee93d139eecd62c41132bbd0ce520"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
