)]}'
{"id":"openstack%2Ftrove~752924","triplet_id":"openstack%2Ftrove~master~I3117cc56aa0edf2f4345784291ae8294a8ffc07a","project":"openstack/trove","branch":"master","topic":"image-build-pip3","hashtags":[],"change_id":"I3117cc56aa0edf2f4345784291ae8294a8ffc07a","subject":"Install pip3 for image build","status":"MERGED","created":"2020-09-21 12:01:46.000000000","updated":"2020-09-21 22:58:21.000000000","submitted":"2020-09-21 22:56:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752924-1600729016347-193fd9bf","meta_rev_id":"8fe21daf7567a67bc87ffc1138498a97f0d2c6f4","_number":752924,"virtual_id_number":752924,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 22:56:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-21 20:42:26.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-21 20:42:26.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 22:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"135613d9534e938d27df53bbbd700856aaa6d208","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-21 12:01:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a287022e9bd480eb27f6b34e4fcad7c091d79ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 14:09:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/507555122d1642d197dd278c99353b0a : FAILURE in 11m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6abff79022284b44806a350eaab7b516 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c94cc6b39554d91aabf01daaa5a875b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb3d5ecb637444349728634aa6a0e4e6 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/670ca9b3336248f5ac2efd15f842a397 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc39e308c6c941229f9205b146a84647 : SUCCESS in 8m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ea0b5e1cfea341b9bd2408361bd94180 : SUCCESS in 10m 01s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0e7dd588181e46948f057428db85dc6c : FAILURE in 1h 58m 02s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd62e1026f884ddeaec19602ec9b9056 : SUCCESS in 1h 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f647e40cd1c0921a0164f2e89559dfc050ddeb1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-21 20:42:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4df497a7b34e22a6265a390c4a1134e18f2889b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 20:42:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41655689222159da59d273cb40f1d4df37807163","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 22:56:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee2d6798754e4c3d807f5cf43cf28501 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b54f3caa804d95a05d847883ac9ea5 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d7c4e430c284331bcd33778bf3cf99e : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3bb52e9241d44e8bc7c711b4f17a008 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f9174136a3f46b8a7664ff7f196699f : SUCCESS in 6m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f8334820c8164abe9c7f1ecd8268872b : FAILURE in 2h 06m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec81632b7913666eb21cc7fd39f15898505b8157","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 22:56:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14dd312e8f4054a7118d341ed51fa038c3948f4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 22:58:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d5b290682a64697a4ad012c005e0543 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7aa269583c1e6bcfa097da58291414e48da24e90","revisions":{"7aa269583c1e6bcfa097da58291414e48da24e90":{"kind":"REWORK","_number":1,"created":"2020-09-21 12:01:46.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/24/752924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/752924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/752924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/752924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/752924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/752924/1"}}},"commit":{"parents":[{"commit":"d75151bbe41b8202385a145f3da023328caf2734","subject":"[Doc] Improve building Trove guest image instruction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d75151bbe41b8202385a145f3da023328caf2734"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-21 12:01:38.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-21 12:01:38.000000000","tz":720},"subject":"Install pip3 for image build","message":"Install pip3 for image build\n\nChange-Id: I3117cc56aa0edf2f4345784291ae8294a8ffc07a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7aa269583c1e6bcfa097da58291414e48da24e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7aa269583c1e6bcfa097da58291414e48da24e90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
