)]}'
{"id":"openstack%2Ftrove~752927","triplet_id":"openstack%2Ftrove~stable%2Fussuri~Ie5a473dcf69de9f7b87dd03987f10239a9bff795","project":"openstack/trove","branch":"stable/ussuri","topic":"ussuri-image-build-pip3","hashtags":[],"change_id":"Ie5a473dcf69de9f7b87dd03987f10239a9bff795","subject":"Install pip3 for image build","status":"MERGED","created":"2020-09-21 12:07:41.000000000","updated":"2020-09-22 01:15:07.000000000","submitted":"2020-09-22 01:13:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752927-1600737190546-c6eda660","meta_rev_id":"fdc3a8ad24607ceea17f91c39e1b969634201dd9","_number":752927,"virtual_id_number":752927,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 01:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-21 20:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-21 20:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 01:13:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa94903488251d1818c9e028a834a2972f8e8eda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-21 12:07:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e55c4da92ceae4c932a460c99fb0920292e69ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 14:37:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8cda4379d174e3fab8c4877c2948c73 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e49ec7962334625b09f8d4fdbec6344 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67b2210c1414607826e8cb19073ed59 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/558123f4fdb649b499dee2a902eddc85 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3208391c651346998348e9f4a8f79c37 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8084b095484cdca348c39ba790f23b : SUCCESS in 6m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2616dd29c6e543948e5c45102e64d0a9 : SUCCESS in 6m 55s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/88294ba21fde4d86835c6ef5bf514a18 : SUCCESS in 47m 09s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6aa9dcb38cdf4ed0ab5eaeec7725b7c7 : SUCCESS in 1h 59m 18s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/e0b9433fdc794296ac7c54a66f4a8f7c : SUCCESS in 2h 22m 10s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2b46f15d366b4719a69f94094ce9b551 : FAILURE in 35m 21s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/e5c1041741994c099310e1fde13aec65 : SUCCESS in 2h 06m 14s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/42b9c1b9a2704a7bba8bcf7d109c20b0 : FAILURE in 41m 17s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/40855c6a6863431fa34e832723e053e2 : FAILURE in 47m 11s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/d888bc9f3dd64ab9bda8620eb34dd820 : FAILURE in 1h 17m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64d7681c117647d099e4844590199d2b : FAILURE in 39m 31s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/cd36c6db4ffd455a925867da209403c8 : SUCCESS in 1h 52m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b5f06b8ded5899cab8586951db73e4440efe180","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-21 20:42:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a810d4ca339c644de4b79af31804263cc7db9d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 23:19:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/010fd4a4e4724c749a546f2323e77125 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/301532214aeb483fb9ee72b5caee2fd2 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50679d4cdcc49b2a41971cb9f890c3f : SUCCESS in 10m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74e8cb490ca04c238a7b54ef029eb601 : SUCCESS in 13m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2eb67b3ad937408c960b655dfb93197a : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c7e7f50e15f4629b9ac7756d1e5b5d2 : SUCCESS in 5m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/61fe4694ad7b4ce296d166e6000eb191 : SUCCESS in 5m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b55ac48c92c24a3d90b989ace570a85c : SUCCESS in 42m 04s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/8827fef42a3e43599f47db2f03b7d996 : SUCCESS in 2h 02m 54s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/5e3a9986357b47e8b6f3899038c0a0a8 : SUCCESS in 2h 18m 39s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/45233a921ef043e5bf139839c45f2e7e : SUCCESS in 1h 21m 39s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/ef6aa57ce32b4aca9aad8e13ad26502c : SUCCESS in 2h 26m 06s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/3966cdaa0f904dbeba57e81e54016c2a : FAILURE in 47m 12s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/afc8e052ec3c47da8769e945929d6907 : FAILURE in 45m 36s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/c769291bcef24b0387764ed9c0ee0408 : FAILURE in 1h 00m 45s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3267e181ffdb486fb7dd38b992701942 : FAILURE in 41m 06s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/4e1d5f35008747e5a53c4af67cf33c01 : SUCCESS in 2h 07m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87cf5bac8b15d4d5d71214e4157d3ef9f5cbf9a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 23:19:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86c674355606aa36ec619de3cbd42cd8b2232b34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 01:13:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"03c7d51e60c71d9bfa9031dc6f2988b1428a07ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 01:13:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af523a4af2bc46fe980e9e192362c08c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c5891460c14424a0ce47f713ce98b4 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34a38d6745514c1a865e13ac7523409d : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e58eb37de0e464c855edfc501de14d8 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a207a52e6554a948739bed24767e6a2 : SUCCESS in 5m 07s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7b5ab54e717745659c2d3af781d4a934 : SUCCESS in 1h 35m 52s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/62ce1780da3a4ce582413f7277980490 : SUCCESS in 1h 51m 25s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/5876c34635cb4615a9b52244172ce06e : SUCCESS in 1h 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f21ea94943a90231e4d5ab0bc45a364b69d9ecf7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 01:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb8b6357a6424b728774385ba754693c : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83bf6ce1b92a292dcf976d506d7cc4635f9909ba","revisions":{"83bf6ce1b92a292dcf976d506d7cc4635f9909ba":{"kind":"REWORK","_number":1,"created":"2020-09-21 12:07:41.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/27/752927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/752927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/752927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/752927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/752927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/752927/1"}}},"commit":{"parents":[{"commit":"f31ac7463a5ee93d139eecd62c41132bbd0ce520","subject":"[Backup] Ignore 404 error when deleting swift objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f31ac7463a5ee93d139eecd62c41132bbd0ce520"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-21 12:07:34.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-21 12:07:34.000000000","tz":720},"subject":"Install pip3 for image build","message":"Install pip3 for image build\n\nChange-Id: Ie5a473dcf69de9f7b87dd03987f10239a9bff795\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83bf6ce1b92a292dcf976d506d7cc4635f9909ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83bf6ce1b92a292dcf976d506d7cc4635f9909ba"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
