)]}'
{"id":"openstack%2Ftrove~753253","triplet_id":"openstack%2Ftrove~master~Ieedbf0c779e2d47aab0f421a7d745a1d7f110c2a","project":"openstack/trove","branch":"master","topic":"limit_fix","hashtags":[],"change_id":"Ieedbf0c779e2d47aab0f421a7d745a1d7f110c2a","subject":"Set default limit for novaclient","status":"MERGED","created":"2020-09-22 08:48:05.000000000","updated":"2020-09-22 23:11:07.000000000","submitted":"2020-09-22 23:09:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753253-1600816169751-193fd9bf","meta_rev_id":"1b38d06b9b866838bd75d382cf77f14396ab2fec","_number":753253,"virtual_id_number":753253,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 23:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-22 20:54:47.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-22 20:54:47.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 20:54:47.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-09-22 23:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a142b22aa1af71be58a7149919ca71fc04847766","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-09-22 08:48:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff8c2e4f71512b3308b273073f231d33f8a463ea","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-22 08:57:21.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"5ddf34c893b59332307ca156f792b6d1421f5578","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-09-22 08:59:58.000000000","message":"Removed reviewer wu.shiming with the following votes:\n\n* Code-Review+1 by wu.shiming \u003cwushiming@yovole.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"6c5b80dfefb8b3709b775bc4410e6150aa56739d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 10:33:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d94eb80bfea647a4bd5ebe95b81dc7d1 : FAILURE in 7m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5033210b7666413fa91291e31f689ec8 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af546b3d15b40389c650b4e831ee7f1 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a62b1e5c5f44e21aac8bc49a113116f : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55972a92c2534274adddecaa4b31cd9a : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979bd40f040c4d31b0b1729b4f8f59a5 : SUCCESS in 6m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b0443c5155c84d088f4cd7707f4b78c0 : SUCCESS in 6m 11s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e5078867f763459fb78469135dc8c759 : FAILURE in 1h 40m 17s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3320b2b8d34c538d47817703a822fc : FAILURE in 1h 09m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b7f24c43b812204797132e12e3f023fbb2c8ab2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-22 20:54:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"31269cff93e215ed4aa63d8b75f6a575bf69ed7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 20:55:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4fc340a42cfcf08dfd634696d5f0b8b65183a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:09:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/498f374158164ec18048821ccf99a26c : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6520eb070de24cd19eacb8b4860d00d5 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59e63f99ecfa4aaa9f57568a69a58c9a : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da4fa555219e418296e802c652512b64 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964cec3a74424e65aca878e3a226525a : SUCCESS in 5m 25s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/812c256f5ee44e529f281c460531d4ec : FAILURE in 2h 06m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10261533abfd931bc87ebdb710697a137b2bdb23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:09:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b12ef4c2f8d52d19fe69adaa2d1c89a97a8d56b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 23:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1ca4baf6244296a36c8bac653d2e2b : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ec38419370c08db78064724d444148cda5b6265","revisions":{"6ec38419370c08db78064724d444148cda5b6265":{"kind":"REWORK","_number":1,"created":"2020-09-22 08:48:05.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/53/753253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/53/753253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/753253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/753253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/53/753253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/53/753253/1"}}},"commit":{"parents":[{"commit":"86a0102a003df890a8d693a60b1e55ba7b0328f1","subject":"Add project_id to backup response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/86a0102a003df890a8d693a60b1e55ba7b0328f1"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-09-22 08:42:27.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-09-22 08:47:53.000000000","tz":480},"subject":"Set default limit for novaclient","message":"Set default limit for novaclient\n\nChange-Id: Ieedbf0c779e2d47aab0f421a7d745a1d7f110c2a\nStory: 2008179\nTask: 40936\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ec38419370c08db78064724d444148cda5b6265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ec38419370c08db78064724d444148cda5b6265"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
