)]}'
{"id":"openstack%2Ftrove~753635","triplet_id":"openstack%2Ftrove~stable%2Fussuri~Id433117ed842177593ae4589cc7d491c760b9e77","project":"openstack/trove","branch":"stable/ussuri","topic":"mgmt-instance-list-stable/ussuri","hashtags":[],"change_id":"Id433117ed842177593ae4589cc7d491c760b9e77","subject":"Set limit to -1 when getting Nova instances","status":"MERGED","created":"2020-09-23 12:51:03.000000000","updated":"2020-09-24 00:01:02.000000000","submitted":"2020-09-23 23:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753635-1600905553107-276e733d","meta_rev_id":"00fc8d1e7c31c26f7d73b2c33b915feeceb9aab3","_number":753635,"virtual_id_number":753635,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-23 23:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-09-23 21:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-09-23 21:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 23:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b780e3951f6bf6c7a7b489f2ff0e66c0cd5231a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-23 12:51:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7578f2b895bcd1c59aaeba70d1ed2975e8e07092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 15:17:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aca0405b405a47bc8bad4b6ac9d1e787 : FAILURE in 7m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1f34ad7f12a4a129771884772c4cba2 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6aa595add74df78ed8644cc2b9504b : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1018843ccf4448ce9ddcff3390766a69 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19e2da1c9c9e4d3e91e3b039572b2ac7 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19b37833f6245fe84dd112f694cfe4a : SUCCESS in 6m 45s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/86ff201678914f449150e6c0aaed776f : SUCCESS in 6m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4755e7d7f6b24b73818cd5f59923fab5 : SUCCESS in 52m 31s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/57350b06bff54a49a45217842ea0f57f : SUCCESS in 1h 52m 15s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ef2cec42a41948a09dd4ed9709fa9005 : SUCCESS in 2h 13m 28s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/67377182a3f94c2a8642a48ab743c157 : SUCCESS in 1h 17m 35s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/b6000105661d43a09e7f3f6f35ce1eeb : SUCCESS in 1h 44m 35s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/5c3acab8803747d5b13599f4f9ac17d7 : FAILURE in 36m 29s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/ccbcf8a633744ee99c43bf06a771e043 : FAILURE in 45m 12s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/76d74ddb93054c24b4fd8c91b6a58593 : FAILURE in 1h 00m 43s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f824e465d5a8474385015f7ece8999ab : FAILURE in 46m 32s (non-voting)\n- trove-functional-mysql-nondev https://zuul.opendev.org/t/openstack/build/b9c4827b013c4dbcbad6eb93a725443d : SUCCESS in 2h 01m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ade82bbf201e990eba70f04c252fcba65a5ad037","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-23 21:49:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da99f20747082ce0d5b2e3092060216476798be5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 21:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"491afd600b9d42944637746a9f30c268cd4fa6c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 23:59:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/239d5a6d17b345b8a626f0563936fc91 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf787780597484a8d4844465cf23a98 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad284f7114ac4f03ad21761d7c8969f9 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/072a95635f0a4da7b51ea2510bb100f7 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cdd0dd53fd481eaf953e25163fefaf : SUCCESS in 5m 06s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e933415159cc419ebab33a21e19cbb98 : SUCCESS in 1h 37m 46s\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/1965273543a64183b04eab7fb2aac234 : SUCCESS in 2h 03m 02s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/2fd75ff5499f4df58aa9c09628db7c02 : SUCCESS in 1h 35m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5a314806787f2d585c8e987c5aa4cc2a6faff1ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 23:59:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8183e605db23db108a9c6fb1f829b70ab110a6d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 00:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be5cf62bb674e30b012be24d89b109d : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e4927086d7db9c6ea6130f94209b00ef8ca0ebd","revisions":{"7e4927086d7db9c6ea6130f94209b00ef8ca0ebd":{"kind":"REWORK","_number":1,"created":"2020-09-23 12:51:03.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/35/753635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/35/753635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/753635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/753635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/35/753635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/35/753635/1"}}},"commit":{"parents":[{"commit":"72049346294ec534d5aa76406ae6f34d109e8983","subject":"Merge \"[Image] Use sudo when removing image cache dir\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72049346294ec534d5aa76406ae6f34d109e8983"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-22 20:59:46.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:51:03.000000000","tz":0},"subject":"Set limit to -1 when getting Nova instances","message":"Set limit to -1 when getting Nova instances\n\nStory: #2008179\nTask: 40936\n\nChange-Id: Id433117ed842177593ae4589cc7d491c760b9e77\n(cherry picked from commit 28453260494076dc5463414b2767188ecb2c1556)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7e4927086d7db9c6ea6130f94209b00ef8ca0ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7e4927086d7db9c6ea6130f94209b00ef8ca0ebd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
