)]}'
{"id":"openstack%2Ftrove~753942","triplet_id":"openstack%2Ftrove~master~I081c53056677f548547509bd820ac2511c383351","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I081c53056677f548547509bd820ac2511c383351","subject":"Only confirm resize when nova instance status is VERIFY_RESIZE","status":"MERGED","created":"2020-09-24 04:17:36.000000000","updated":"2020-10-01 06:38:48.000000000","submitted":"2020-10-01 06:36:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753942-1601534200986-2794df33","meta_rev_id":"6b15340bb2e478372f7a791badfcde67bdb6bf61","_number":753942,"virtual_id_number":753942,"owner":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 06:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-01 03:14:10.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-01 03:14:10.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 03:14:10.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-10-01 06:36:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"513b78d31440af2786fb6f1fac381769d9f17a12","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-09-24 04:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bce635114246e8180e5846b7b92b8fb643137c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:38:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9794551da7404ae1aeb61cafa67c4c50 : FAILURE in 6m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f31aebe81370419db3d48c9c6ca32e83 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd23bb125c134b3b807f99a1494ed965 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04c4fd560ae248b4ac144918eb32d448 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ad599614dda40cbb1e3460aeb03ec91 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7a77f49f3b483888689a86f14e77fa : SUCCESS in 6m 06s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d834f3b589894077930d2f96b62a6cf2 : SUCCESS in 5m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b098d505167d4fa3b8b99cc367820e0b : TIMED_OUT in 2h 13m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/365f651632e4444e814c0de6d8e73b35 : SUCCESS in 1h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71ab24dd57c5ada7523d7609a068013e260ed6a2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 19:37:26.000000000","message":"Patch Set 1: Code-Review-1\n\nThe recommended way is to add a check for VM status before https://github.com/openstack/trove/blob/master/trove/taskmanager/models.py#L1839","accounts_in_message":[],"_revision_number":1},{"id":"bd373b29e40610ddd7e7b0b42e08086c2d758251","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-09-25 01:40:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01414c26a724b6145fbc60a25c8b4fedcb926489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 03:49:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39d5c1ef92f8420b99083cdad3d1531d : FAILURE in 7m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e80174c7d3ac461fa1dad90ee71c633a : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec854d72a40b473b93935aea0f7a1c69 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eecd8bebe5d042d692618b35fafe3767 : FAILURE in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b3aeed7d6294867ba2d5f85f640e7dc : FAILURE in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7edcad02b0054a538193dc4fa1b6a0c9 : SUCCESS in 5m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/09b0d933f66447959f796383cf0e4f21 : SUCCESS in 6m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4d66041c13cf418e9baf1cefd3a22bba : FAILURE in 2h 02m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2e2bce9bf384b5da6fe210e3956e871 : SUCCESS in 1h 06m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"665ad90168838f10383f4fb2ec25772ef46d1176","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-09-25 04:33:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5a1b239411200de226951635815b3e7c5c0e072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:44:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2398929266e4852a05027c9cd415b20 : FAILURE in 6m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e04f262faa9f4cf895218488f2ff65ff : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62dde146f79a47d6b4d5142bc45f769d : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14cb5c8a4a8a4a188a45168eb35e0e5d : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b41dfe83c1ab43a88ff9ce810866d5d0 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6057199a8ee1481a9805acb73cd9adab : SUCCESS in 5m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eeab6ce58a3c4789a4c63bdee9f7009a : SUCCESS in 4m 56s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/83d1d260ab5b4ccbb2f94bfd62ff3fe6 : FAILURE in 2h 09m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62819473c26b4576ad70834f3ce56c08 : SUCCESS in 1h 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"13f70d25f6499772ec2fe4248bd5b9e8e8d08c69","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-30 12:22:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0eea32c57cdc97d424aedee1ce7eba5d1580d3bf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-30 12:22:50.000000000","message":"Patch Set 3:\n\nCode looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"565791695313aa04ecde65b1433648cce5972586","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-10-01 02:21:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4036276995013a37c709535e2668ef90ef8df3a4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 03:14:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ab516a5a66a2064166e800a31683386ebeba00e7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 03:14:43.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 2101194849d50c2384489527ddf12a51aebeaa48","accounts_in_message":[],"_revision_number":4},{"id":"acb9d655fcd5dac3a05984ab3b403eaf2a751d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 04:35:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6495c2c804d040e6a901a716e8071c32 : FAILURE in 6m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da4ee0f0566c4a1f969e2d60740d1e66 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e74abef4aae442aba606fd2ec3750cd : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/412a19f80d3244eda89c239df6673e90 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7a1bed6bab446ee8e844ebf39b071e0 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533abcb86e544d47aae93c86c2fa64c6 : SUCCESS in 5m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a6fd1189feb45558a1be619d28ad2f9 : SUCCESS in 5m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/20410a086b4b451d85c29d8870fa559e : TIMED_OUT in 2h 12m 47s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd3710ab68fe46deb90a407ee81f1358 : FAILURE in 1h 44m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"73ad1f453cda58e5a6332b059c464bcc30cefb57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 04:35:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a7d490a767106658cb16fd5489784d18db95585d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 06:36:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"463b64a7be428cb5fe4b3b394fa9ec5c73e31f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 06:36:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdd96fbc58464d3680b9b3c65e8ead6f : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f2f533d8c64489b2966258afb1de1b : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d0718686cee4c0aa5d8f5808548feed : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef31bdecf856477ba31169a9719a1591 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4431cf3698734c76a5e3b44178e48156 : SUCCESS in 5m 03s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8838f63771e84bee9ec715e3001da6f8 : FAILURE in 1h 55m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6b15340bb2e478372f7a791badfcde67bdb6bf61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 06:38:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3d7c0e60eb4bf58fc84bd82f67b7d6 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c46bd0615a824de1a062965cdd4f7b73ea567563","revisions":{"806bf11754827f6e86e6f7ab6162f22f52a85141":{"kind":"REWORK","_number":1,"created":"2020-09-24 04:17:36.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/42/753942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/753942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/753942/1"}}},"commit":{"parents":[{"commit":"dc2faefa720d08ce77868fade696dd93f548746d","subject":"Merge \"Set limit to -1 when getting Nova instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dc2faefa720d08ce77868fade696dd93f548746d"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-09-24 04:14:27.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-09-24 04:14:27.000000000","tz":420},"subject":"Pass confirm resize when resize database flavor","message":"Pass confirm resize when resize database flavor\n\nWhen nova compute config auto confirm resize, resize database flavor will fail\nbecause nova vm was activated. It raise exception 409 Conflict. We can ignore\nthis exception\n\nChange-Id: I081c53056677f548547509bd820ac2511c383351\nStory: #2008189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/806bf11754827f6e86e6f7ab6162f22f52a85141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/806bf11754827f6e86e6f7ab6162f22f52a85141"}]},"branch":"refs/heads/master"},"e2ba590b5b9edcfd6692ac1ae262654fdd75b4d0":{"kind":"REWORK","_number":2,"created":"2020-09-25 01:40:13.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/42/753942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/753942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/753942/2"}}},"commit":{"parents":[{"commit":"957e0c1adf1196d8e2fdd5428576bb124d95ddae","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/957e0c1adf1196d8e2fdd5428576bb124d95ddae"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-09-25 01:38:37.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-09-25 01:38:37.000000000","tz":420},"subject":"Only confirm resize when nova instance status is VERIFY_RESIZE","message":"Only confirm resize when nova instance status is VERIFY_RESIZE\n\nWhen nova compute config auto confirm resize, resize database flavor will\nfail because nova vm was activated. It raise exception 409 Conflict.\nWe just only confirm resize when nova instance status is VERIFY_RESIZE\n\nChange-Id: I081c53056677f548547509bd820ac2511c383351\nStory: #2008189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e2ba590b5b9edcfd6692ac1ae262654fdd75b4d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e2ba590b5b9edcfd6692ac1ae262654fdd75b4d0"}]},"branch":"refs/heads/master"},"3ff004d706d7ead66abbfacfd651cc548d66a702":{"kind":"REWORK","_number":3,"created":"2020-09-25 04:33:49.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/42/753942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/753942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/753942/3"}}},"commit":{"parents":[{"commit":"957e0c1adf1196d8e2fdd5428576bb124d95ddae","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/957e0c1adf1196d8e2fdd5428576bb124d95ddae"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-09-25 01:38:37.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-09-25 04:33:33.000000000","tz":420},"subject":"Only confirm resize when nova instance status is VERIFY_RESIZE","message":"Only confirm resize when nova instance status is VERIFY_RESIZE\n\nWhen nova compute config auto confirm resize, resize database flavor will\nfail because nova vm was activated. It raise exception 409 Conflict.\nWe just only confirm resize when nova instance status is VERIFY_RESIZE\n\nChange-Id: I081c53056677f548547509bd820ac2511c383351\nStory: #2008189\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff004d706d7ead66abbfacfd651cc548d66a702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3ff004d706d7ead66abbfacfd651cc548d66a702"}]},"branch":"refs/heads/master"},"c46bd0615a824de1a062965cdd4f7b73ea567563":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2020-10-01 02:21:19.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/42/753942/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/42/753942/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/42/753942/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/42/753942/4"}}},"commit":{"parents":[{"commit":"2f336f3eb1d146b416d39504d603640254d7ef3f","subject":"[Postgresql] Adding config parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f336f3eb1d146b416d39504d603640254d7ef3f"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-10-01 02:20:18.000000000","tz":420},"committer":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-10-01 02:21:15.000000000","tz":420},"subject":"Only confirm resize when nova instance status is VERIFY_RESIZE","message":"Only confirm resize when nova instance status is VERIFY_RESIZE\n\nWhen nova compute config auto confirm resize, resize database flavor will\nfail because nova vm was activated. It raise exception 409 Conflict.\nWe just only confirm resize when nova instance status is VERIFY_RESIZE\n\nChange-Id: I081c53056677f548547509bd820ac2511c383351\nStory: #2008189\nTask: #40950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c46bd0615a824de1a062965cdd4f7b73ea567563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c46bd0615a824de1a062965cdd4f7b73ea567563"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
