)]}'
{"id":"openstack%2Ftrove~755408","triplet_id":"openstack%2Ftrove~master~Ica4f888bf1bc7b771cdfcb0af5c104b2aa6410fe","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ica4f888bf1bc7b771cdfcb0af5c104b2aa6410fe","subject":"Handle case where no networks are selected at create","status":"MERGED","created":"2020-09-30 23:11:27.000000000","updated":"2020-10-01 10:56:59.000000000","submitted":"2020-10-01 10:55:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755408-1601549720044-e452fb22","meta_rev_id":"fa03633a9d7d725bcd6f749b8aa43b330e20c3e0","_number":755408,"virtual_id_number":755408,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 10:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-01 08:36:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-01 08:36:11.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 08:36:11.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-10-01 10:55:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d123ce465db069f4ad9a909dfa1bd1f2cacb494","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-30 23:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6da6747a908f6fda7d04341e43588dd1c5537c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 00:56:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a70a8b35996429092755372dd8eb54a : FAILURE in 6m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f80c405b9f54a60a6ce455274e75733 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13cd76fd91024c848f51f4f3ffcd30e7 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd757e7b60f24b3b998577ae1d2c7995 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d6b04e95bbc4130bf84218f9a77632e : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bceba212fe80430ca593a8374265ba11 : SUCCESS in 5m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a2f9d6cbe05543c5b399fbac0926e1da : SUCCESS in 6m 31s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6908e5c017ab4d098e46375dca74aa72 : FAILURE in 1h 41m 26s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/feedccf91dfc4b77a24c4640494a915e : FAILURE in 1h 13m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"397a618cb65193d0bbf89d50420679959d4a6015","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 08:36:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"81742c461bbf59f91e954055cc905846d18579cf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 08:36:20.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 6ccdb4b82f1f7f895ca30972bbd359fa5e0c5659","accounts_in_message":[],"_revision_number":1},{"id":"d81fbc094f157cc52fc93314551d3b94c34a5558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:36:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e15cf45acde3523b6b469ec1b1aee37a8e31491","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 08:36:40.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit a5a5c5e060c37c4d0ec5fff125253ca02c1f1817","accounts_in_message":[],"_revision_number":1},{"id":"d1413dbea9e54e8e11c5604a49a94ef5ed65a87a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:55:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fad2f5a2ea064fcc95ee0a136ae811d8 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9933dd18794b10b3d27804180d86a1 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca8e7a5513b547b39096ffa1b5229e71 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f565454f8645a78d702a71d3b27ae9 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86362bf9296546c5a260ec85d81cb19d : SUCCESS in 6m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f7d788ddd39e4387951f8d485afe5289 : TIMED_OUT in 2h 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8ee96c0f779898fecf19ae5571ea2d167508f0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:55:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa03633a9d7d725bcd6f749b8aa43b330e20c3e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:56:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf680199b9d439793c7fbdce7c89f09 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9a99bc483dec34677d069a933afce400a73d0a1","revisions":{"f9a99bc483dec34677d069a933afce400a73d0a1":{"kind":"REWORK","_number":1,"created":"2020-09-30 23:11:27.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/08/755408/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/08/755408/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/755408/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/755408/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/755408/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/08/755408/1"}}},"commit":{"parents":[{"commit":"b9cdf71757ed68cbac1c02e6631a151a82dd1bc4","subject":"Merge \"Remove six.moves.urllib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9cdf71757ed68cbac1c02e6631a151a82dd1bc4"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-30 23:06:53.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-30 23:06:53.000000000","tz":600},"subject":"Handle case where no networks are selected at create","message":"Handle case where no networks are selected at create\n\nCurrently trove will error when no networks are selected\nand no management network is configured. This change will\npass in None as the nics to nova server create which will\nlet nova choose the network.\n\nChange-Id: Ica4f888bf1bc7b771cdfcb0af5c104b2aa6410fe\nStory: 2008213\nTask: 40999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9a99bc483dec34677d069a933afce400a73d0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f9a99bc483dec34677d069a933afce400a73d0a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
