)]}'
{"id":"openstack%2Ftrove~755426","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~I081c53056677f548547509bd820ac2511c383351","project":"openstack/trove","branch":"stable/victoria","hashtags":[],"change_id":"I081c53056677f548547509bd820ac2511c383351","subject":"Only confirm resize when nova instance status is VERIFY_RESIZE","status":"MERGED","created":"2020-10-01 03:14:43.000000000","updated":"2020-10-01 10:57:02.000000000","submitted":"2020-10-01 10:55:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755426-1601549725482-c1ef629e","meta_rev_id":"14cb6cdeb5ee4d47a7fb03fb225986bca02408d7","_number":755426,"virtual_id_number":755426,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 10:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-01 08:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-01 08:44:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-01 03:14:43.000000000","_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 03:14:43.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"state":"REVIEWER"},{"updated":"2020-10-01 10:55:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5ec34ff42b41c6c340c7fd98e97d6e286ec64a2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 03:14:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fdeb5464f91fb7e6aeb3aede18cfba8543e86228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 05:37:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6348bbd24bb42eab86d2728b607ee47 : FAILURE in 7m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bae661fc1c34d5f95c436b4dc64ef2c : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd37a6fa0aa14fe4b0fdc3b4df988755 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b12f4c12c70464c8da54695442a309e : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bc1c24f318e4516a6ab84b42ede39df : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab32be73ff64bc9bb37f529e9b88fd4 : SUCCESS in 7m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21904e7bbc3b45468b0b4ac510a5e6f3 : SUCCESS in 6m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dfc0fc93223e4cbbbd8b06acbc9ff62c : TIMED_OUT in 2h 13m 00s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81a042f4559446a0b8668a9b6bd64e86 : FAILURE in 1h 08m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2efed15a1ff255d06799b2e87f764df26153db5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 08:44:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3eac50d3b372caad788883379485b66fe45a796c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 08:45:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66a7f0d5c4c74304183730f14f793b7a15d16261","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:55:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed6e6804d674ba7a5ac5d13c0c83ba5 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26938d68c10d4eb08bf9c45ca92590ab : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a03480c6b6114d4e99d408d7db057374 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c4e3a85dc3d49229c3eeee2a0a2b78c : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723f7c1d507548138fe5a6b662bde6e7 : SUCCESS in 5m 31s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d6738c09a4db45a196629b77d377b302 : FAILURE in 1h 46m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e0e22600dc55cf417039ad649da7855de855842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:55:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14cb6cdeb5ee4d47a7fb03fb225986bca02408d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da10b96862040d8b9407632ff1dc9e3 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2101194849d50c2384489527ddf12a51aebeaa48","revisions":{"2101194849d50c2384489527ddf12a51aebeaa48":{"kind":"REWORK","_number":1,"created":"2020-10-01 03:14:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/26/755426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/26/755426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/755426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/755426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/26/755426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/26/755426/1"}}},"commit":{"parents":[{"commit":"ac664f2382a95e38274b7289af930c9c90714171","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ac664f2382a95e38274b7289af930c9c90714171"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-10-01 02:20:18.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-01 03:14:43.000000000","tz":0},"subject":"Only confirm resize when nova instance status is VERIFY_RESIZE","message":"Only confirm resize when nova instance status is VERIFY_RESIZE\n\nWhen nova compute config auto confirm resize, resize database flavor will\nfail because nova vm was activated. It raise exception 409 Conflict.\nWe just only confirm resize when nova instance status is VERIFY_RESIZE\n\nChange-Id: I081c53056677f548547509bd820ac2511c383351\nStory: #2008189\nTask: #40950\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2101194849d50c2384489527ddf12a51aebeaa48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2101194849d50c2384489527ddf12a51aebeaa48"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
