)]}'
{"id":"openstack%2Ftrove~755427","triplet_id":"openstack%2Ftrove~master~I6ba7ea392df3b51ed9d230cacab14869348c217c","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I6ba7ea392df3b51ed9d230cacab14869348c217c","subject":"Add user/project/trove-id metadata to trove server instances","status":"MERGED","created":"2020-10-01 04:02:56.000000000","updated":"2020-10-02 11:51:45.000000000","submitted":"2020-10-02 11:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755427-1601639417594-8f1aa1ae","meta_rev_id":"89a1b4cf6ac762ad00238fdd06243cb57976129e","_number":755427,"virtual_id_number":755427,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 11:50:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":2,"date":"2020-10-02 09:32:41.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":1,"date":"2020-10-02 09:32:41.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 04:02:56.000000000","updated_by":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"reviewer":{"_account_id":18418,"name":"Andy Botting","email":"andy@andybotting.com","username":"andybotting"},"state":"REVIEWER"},{"updated":"2020-10-02 09:32:41.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-10-02 11:50:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e91cb56975421812951bdc269028659437f261cb","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-01 04:02:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e845a6a7b696820d901538f043bc0f475a2a7bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 06:03:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80da799ec17a466dad92fc85328a0d54 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/559fdd8b19344e29ac8f803d9c88e8c9 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d4b2e689674e72910f9041c01ca50e : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11a2834c1d334976bdf9b84577ca7ffc : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8f31f06657d4e2eb79b7e45ba41439b : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82343ff6a6114d98958e849dc4366c26 : SUCCESS in 5m 56s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/677e1a0843f54b7caa23ce31d3db7628 : SUCCESS in 5m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/56f1a3bcd0d2455e90a3396cf530a7ea : FAILURE in 1h 54m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0de39c2642045ee9a5cec1261b987e6 : SUCCESS in 1h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7a076752e0858363ca2cdb3d1e8c76eb6bd3a98","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 08:44:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9711f68338d8a8dfaee5b9d20c277069957361e1","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-02 01:16:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02bea59b433ee6477c065f72d0a3f05de24a2c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 03:36:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/699b8870265446d48b2493432678ab6b : FAILURE in 6m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad073cade71a4a969588ad5ce001efc7 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2077d07249e1404fb4ae6a1c3017c653 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be0ac71d93a641e489f58bf3433cd8ba : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2271f2730694ef6a89d6dad93aa136d : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53b3684ce9244b4a524bb1e1a29e08d : SUCCESS in 5m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff005002c5a0487da6d87d255e500960 : SUCCESS in 5m 54s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/11deb4d7c691442bb6b34864c122da84 : TIMED_OUT in 2h 12m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f80b15c015294f8fbd79ec41ab051909 : FAILURE in 1h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ad89b0ffa3451a0a518ceccbd05a60a205a0dde5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-02 09:32:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthx","accounts_in_message":[],"_revision_number":2},{"id":"256c86c248acb5eef035ebf0fbcec35816e2ece7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:32:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d845825ec3718c0deff846becc8d45befac99d1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:50:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dbdedbb87e0af36fbd090b9b9d0ba0e62d60a6a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:50:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67d2ce07d3544ba1acfae4773fa2c48e : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473a0164b959446695e427d93cec9e63 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62b17d8f482846e4acfe87f49807e5da : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e782fd13dc64377a1a8a2203d9dceb3 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40cdfe5badb49c5b5e5fd41e96600e7 : SUCCESS in 5m 46s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/387d86c8351d4e8e8f7ca949bd39ce6c : FAILURE in 2h 09m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89a1b4cf6ac762ad00238fdd06243cb57976129e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 11:51:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e644a01ce66498d807f3302fa50d5cf : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"adc29948faa4e033e10dbc7c440427437cd555a3","revisions":{"3bca077417f35c49f529269cc938a8addfd494df":{"kind":"REWORK","_number":1,"created":"2020-10-01 04:02:56.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/27/755427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/755427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/755427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/755427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/755427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/755427/1"}}},"commit":{"parents":[{"commit":"2f336f3eb1d146b416d39504d603640254d7ef3f","subject":"[Postgresql] Adding config parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2f336f3eb1d146b416d39504d603640254d7ef3f"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-08-25 01:33:46.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-01 04:01:57.000000000","tz":600},"subject":"Add user/project/trove-id metadata to trove server instances","message":"Add user/project/trove-id metadata to trove server instances\n\nThis can be handy when wanting to trace back what trove instance\nand real user/project the nova instance belongs too.\n\nChange-Id: I6ba7ea392df3b51ed9d230cacab14869348c217c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bca077417f35c49f529269cc938a8addfd494df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bca077417f35c49f529269cc938a8addfd494df"}]},"branch":"refs/heads/master"},"adc29948faa4e033e10dbc7c440427437cd555a3":{"kind":"REWORK","_number":2,"created":"2020-10-02 01:16:56.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/27/755427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/755427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/755427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/755427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/755427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/755427/2"}}},"commit":{"parents":[{"commit":"af83edd1c44d0991344a5f6b6746d039c1515554","subject":"Merge \"Handle case where no networks are selected at create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/af83edd1c44d0991344a5f6b6746d039c1515554"}]}],"author":{"name":"Andy Botting","email":"andy@andybotting.com","date":"2017-08-25 01:33:46.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-02 01:16:43.000000000","tz":600},"subject":"Add user/project/trove-id metadata to trove server instances","message":"Add user/project/trove-id metadata to trove server instances\n\nThis can be handy when wanting to trace back what trove instance\nand real user/project the nova instance belongs too.\n\nChange-Id: I6ba7ea392df3b51ed9d230cacab14869348c217c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/adc29948faa4e033e10dbc7c440427437cd555a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/adc29948faa4e033e10dbc7c440427437cd555a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
