)]}'
{"id":"openstack%2Ftrove~755447","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~Ica4f888bf1bc7b771cdfcb0af5c104b2aa6410fe","project":"openstack/trove","branch":"stable/victoria","hashtags":[],"change_id":"Ica4f888bf1bc7b771cdfcb0af5c104b2aa6410fe","subject":"Handle case where no networks are selected at create","status":"MERGED","created":"2020-10-01 08:36:20.000000000","updated":"2020-10-01 22:32:48.000000000","submitted":"2020-10-01 22:31:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755447-1601591484907-c7f4b8dc","meta_rev_id":"fcc719ece37afde861b9c7cb47cd8b52583621c6","_number":755447,"virtual_id_number":755447,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 22:31:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2020-10-01 20:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"date":"2020-10-01 08:36:20.000000000","_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2020-10-01 20:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 08:36:20.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-10-01 22:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7e958fc8d42f420f8b300147fe54d5e157182e1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 08:36:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8049222117b0a68807383e83505d9e0b3c4b3f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 10:54:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e01ae4e3e2247abba31c043831c2aed : FAILURE in 6m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f8125c05b04192b62839ce1f527e2d : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6cd68058fc44a4a92deb1182a9f5807 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cab63d847e3490cbf8a5828d7331943 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb9917c96014f0180e01ba554179a2b : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e692f12ecc154a6ea0a7e5e54b7a23e8 : SUCCESS in 6m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48edf5f1ca0d4ee38fe2dc0332a61f52 : SUCCESS in 6m 27s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/77827d037fbc45fc9ab63b5088927a55 : TIMED_OUT in 2h 13m 08s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3741488c89a4829bb6cd774fa6be61f : FAILURE in 1h 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c783a602246a1a36f2ad6f0a919d16ac50f6367","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-01 20:48:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbe67de9be2b62999694b333b0cf7b5157b36e0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 20:48:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4b72ade9ea0ee96d4a8eacb50c7669823bccbee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:31:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b8122e709dd4b988b799e133414a1ce : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/323e93f6d4f4472aa28a858c59160898 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5361d2790ce9474f8e0aefffb1be042d : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/389107d4369c4156b54da574189aed65 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7919517f273a49049e411c1483366d7c : SUCCESS in 6m 40s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/931867e32be645769c0c442a18b0ebbc : FAILURE in 1h 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a806b2bf4716de27d499f55e69ca798a28f1e869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:31:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fcc719ece37afde861b9c7cb47cd8b52583621c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 22:32:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a161961220c245388d8d3c0e5405ad43 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ccdb4b82f1f7f895ca30972bbd359fa5e0c5659","revisions":{"6ccdb4b82f1f7f895ca30972bbd359fa5e0c5659":{"kind":"REWORK","_number":1,"created":"2020-10-01 08:36:20.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/47/755447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/755447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/755447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/755447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/755447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/755447/1"}}},"commit":{"parents":[{"commit":"aa3c53e63604c283d82417139b9ca02ffc6db222","subject":"[Postgresql] Adding config parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/aa3c53e63604c283d82417139b9ca02ffc6db222"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-09-30 23:06:53.000000000","tz":600},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-01 08:36:20.000000000","tz":0},"subject":"Handle case where no networks are selected at create","message":"Handle case where no networks are selected at create\n\nCurrently trove will error when no networks are selected\nand no management network is configured. This change will\npass in None as the nics to nova server create which will\nlet nova choose the network.\n\nChange-Id: Ica4f888bf1bc7b771cdfcb0af5c104b2aa6410fe\nStory: 2008213\nTask: 40999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ccdb4b82f1f7f895ca30972bbd359fa5e0c5659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6ccdb4b82f1f7f895ca30972bbd359fa5e0c5659"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
