)]}'
{"id":"openstack%2Ftrove~756870","triplet_id":"openstack%2Ftrove~master~I0e51d08515c121e3a7b0e82b6e4c4161bb4fbc4a","project":"openstack/trove","branch":"master","topic":"image-tag","hashtags":[],"change_id":"I0e51d08515c121e3a7b0e82b6e4c4161bb4fbc4a","subject":"Image tags support in datastore version","status":"MERGED","created":"2020-10-08 15:37:12.000000000","updated":"2020-10-09 02:11:07.000000000","submitted":"2020-10-09 02:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756870-1602209331739-6cf76486","meta_rev_id":"25054faede8673506e3321880e8471b44911120c","_number":756870,"virtual_id_number":756870,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 02:08:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-09 01:17:41.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-09 01:17:41.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 02:08:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"328c01330cec32809e27995ae370e0720c21f0cc","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-08 15:37:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c80922f70970a3b4d988ad2f3f5ceff4aa93e83a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 17:51:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fc6c6695aa014a098678257580ff13ed : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/788a3ecf9f3a4c5e9cabc909d54df960 : FAILURE in 7m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34edccfe591244f6855675f9329d19fc : FAILURE in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd61c7f752e44de4afbb412e2820f17e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dc377751c534fbeb6e2ab661654b37a : FAILURE in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99d7a133d27d4c93970f76206696a0b4 : FAILURE in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e6369e57ff4d37b0df515d943dddb0 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77e1a06b3ec540bcba31e238f74f76e5 : SUCCESS in 7m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ade8250c7c1b49ffae9585a45d7e844a : SUCCESS in 6m 19s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1b061a2fbbb74d139374db21b9f3f78b : FAILURE in 1h 59m 15s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/743c5ebba9f44fabab8514bb9cc326c2 : FAILURE in 1h 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"584a8e973e5b04430182eeef089b5d65bc4a96ee","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-08 22:46:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3093e2b20d7bb0ef35348237a53ced5c823dd7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 01:06:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7887076e5642487b8657715c899a844a : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64d402ea2b5442639fbc3c16e905affc : FAILURE in 6m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/970e5acb279b430bb8ad61377e9ddd7c : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc705b3096da4fc8b3cde5b9d1b518d5 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12dd62e086ed4930b1d83225a2b94999 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2810606a27a46f2b2b33c503718882d : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f16d08761448c8bccbacefbf20e662 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f069a9f1fb17421480fb3acc1d3bdc3c : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63cde811799e420d8fdb9b3b6c5fc72c : SUCCESS in 5m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/fea8c7292d44453a881509d9663e3100 : TIMED_OUT in 2h 12m 58s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18f2ab0226bc4887ade95ede81db709e : FAILURE in 1h 48m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b40dd36aad9d1b1eaf117e2878ef3b3f192c155a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-09 01:17:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"060b23ae281953bbe142aedf5835156f10fe2dcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 01:17:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d9b9c97460ef19142692b2ed4ec057637ac29639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:08:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a219a83c2c8e455cb6ff18f52669354e : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7010a296962343728f71c518bfaaf3a5 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c83d279f2da44cc3a6974d153f5143fa : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/697579b3e3c14aba85d63a764bcad371 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af47c132f71a4a58b46e157a8a93498a : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01065fd72a0941b5880178aa676d57e8 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ef198cfe4564aaab39f4aca5641e735 : SUCCESS in 5m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8b4dbdcd01e8403c863df525007e5eb6 : FAILURE in 48m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bccb8b08afb3a285ca5a61cb398fa68f363ad57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:08:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"64b8b16c8175a6bdba85a534f9ee13186b6d62ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:11:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/11bcb6b7d0cb445d9112fbc17fbd214d : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8077858ca2434fb8e087cc64789a0a : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60d4c00e6ba471899729878f208bf61 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d24b65052bbfb8f7dff5744c7bb44a58a2336c6","revisions":{"72fd6b41d8b8a66e00cf4731fd97f6b6c57b42ef":{"kind":"REWORK","_number":1,"created":"2020-10-08 15:37:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/70/756870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/756870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/756870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/756870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/756870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/756870/1"}}},"commit":{"parents":[{"commit":"680c7b8361521c494ae93d13b5d33c43a27d35ad","subject":"Set default-authentication-plugin default for mysql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/680c7b8361521c494ae93d13b5d33c43a27d35ad"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-07 05:50:53.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-08 12:49:57.000000000","tz":780},"subject":"Image tags support in datastore version","message":"Image tags support in datastore version\n\nChange-Id: I0e51d08515c121e3a7b0e82b6e4c4161bb4fbc4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72fd6b41d8b8a66e00cf4731fd97f6b6c57b42ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/72fd6b41d8b8a66e00cf4731fd97f6b6c57b42ef"}]},"branch":"refs/heads/master"},"1d24b65052bbfb8f7dff5744c7bb44a58a2336c6":{"kind":"REWORK","_number":2,"created":"2020-10-08 22:46:13.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/70/756870/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/70/756870/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/756870/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/756870/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/70/756870/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/70/756870/2"}}},"commit":{"parents":[{"commit":"680c7b8361521c494ae93d13b5d33c43a27d35ad","subject":"Set default-authentication-plugin default for mysql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/680c7b8361521c494ae93d13b5d33c43a27d35ad"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-07 05:50:53.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-08 22:43:48.000000000","tz":780},"subject":"Image tags support in datastore version","message":"Image tags support in datastore version\n\nChange-Id: I0e51d08515c121e3a7b0e82b6e4c4161bb4fbc4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d24b65052bbfb8f7dff5744c7bb44a58a2336c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1d24b65052bbfb8f7dff5744c7bb44a58a2336c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
