)]}'
{"id":"openstack%2Ftrove~757976","triplet_id":"openstack%2Ftrove~master~I0ef055f6f031e43ab7deb46fd56112b5c49f6f4c","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I0ef055f6f031e43ab7deb46fd56112b5c49f6f4c","subject":"Remove six from requirements","status":"ABANDONED","created":"2020-10-14 01:30:53.000000000","updated":"2020-10-16 02:46:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5c09de844a026746bc3324dceffe62331197883","_number":757976,"virtual_id_number":757976,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-15 22:18:04.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-14 03:35:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-16 02:39:20.000000000","_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 03:35:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 22:18:04.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"83e4a3d86d31a0f24c38d51e23402e4590d8afda","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-10-14 01:30:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4381ab0b0e7a230f5317da2cd6d7e0bd89e76b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 03:35:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2da5454e6e234161aa258301d394fc62 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25b26168a74c4f578068cb9757adf599 : FAILURE in 6m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f151b20cd8c45a88386be541e1d57cc : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ba8e16b19e41e1aac1d3bb74271c01 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f9ab18529641358b3b63cfb66121af : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb1989e27034ffba1252ce6f5d458b5 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7858799602d740a6bc30ea38a9a68ac5 : SUCCESS in 5m 54s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3e547ae8a3674e64aa848ce4e9bd92b8 : FAILURE in 2h 00m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0471c77d062f4b5c9bcbaa2c048095f7 : FAILURE in 1h 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1c08db00440a85c08a8051c81c14579dcd8be1c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-14 07:37:26.000000000","message":"Patch Set 1:\n\nAny particular reason the six removals are split into multiple patches?","accounts_in_message":[],"_revision_number":1},{"id":"91e8a187a5124c6e86a2864b4ea9a47d32494302","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-10-15 08:18:58.000000000","message":"Patch Set 1:\n\n\u003e Any particular reason the six removals are split into multiple\n \u003e patches?\n\nBecause there are too many files, I separate the test from other parts when modifying locally.\nThe basestring part is separated to make the review clearer.","accounts_in_message":[],"_revision_number":1},{"id":"e5b900c8fa11eb64b0f5e90958928747c11cda64","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-15 22:17:01.000000000","message":"Patch Set 1:\n\n\u003e \u003e Any particular reason the six removals are split into multiple\n \u003e \u003e patches?\n \u003e \n \u003e Because there are too many files, I separate the test from other\n \u003e parts when modifying locally.\n \u003e The basestring part is separated to make the review clearer.\n\nGiven the change is simple, please just merge them into one single commit to make it easier to track and merge.","accounts_in_message":[],"_revision_number":1},{"id":"b0bb4fd89aadde98d9d67266275fc70e5f5aed98","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-15 22:18:04.000000000","message":"Patch Set 1:\n\nWhat I want to avoid is too many similar commits for such non-feature patches.","accounts_in_message":[],"_revision_number":1},{"id":"96edb348227a0a7bcd562eb3d384134bf8d0eaf3","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-10-16 02:39:20.000000000","message":"Patch Set 1:\n\n\u003e What I want to avoid is too many similar commits for such\n \u003e non-feature patches.\n\nI got it.","accounts_in_message":[],"_revision_number":1},{"id":"c5c09de844a026746bc3324dceffe62331197883","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-10-16 02:46:32.000000000","message":"Abandoned\n\nmerge to https://review.opendev.org/#/c/758516/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f281539f0f696e603ad9f97d41bb03365ed4ec6","revisions":{"0f281539f0f696e603ad9f97d41bb03365ed4ec6":{"kind":"REWORK","_number":1,"created":"2020-10-14 01:30:53.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/76/757976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/76/757976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/757976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/757976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/76/757976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/76/757976/1"}}},"commit":{"parents":[{"commit":"9e2515aad353c0ad6ced3558c6cc0fcd6d478fff","subject":"fix typo in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9e2515aad353c0ad6ced3558c6cc0fcd6d478fff"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-10-14 01:23:22.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-10-14 01:23:22.000000000","tz":480},"subject":"Remove six from requirements","message":"Remove six from requirements\n\nRemove six from following files.\n\n- ubuntu-requirements.txt\n- lower-constraints.txt\n- requirements.txt\n\nDepends-On: https://review.opendev.org/#/c/757771/\nDepends-On: https://review.opendev.org/#/c/757783/\nDepends-On: https://review.opendev.org/#/c/757792/\nChange-Id: I0ef055f6f031e43ab7deb46fd56112b5c49f6f4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f281539f0f696e603ad9f97d41bb03365ed4ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0f281539f0f696e603ad9f97d41bb03365ed4ec6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
