)]}'
{"id":"openstack%2Ftrove~758651","triplet_id":"openstack%2Ftrove~master~I4cfddd6085f0ce4605585563a3e043b2d85ad5b7","project":"openstack/trove","branch":"master","topic":"typo","hashtags":[],"change_id":"I4cfddd6085f0ce4605585563a3e043b2d85ad5b7","subject":"Fix a typo in the document","status":"ABANDONED","created":"2020-10-17 10:24:28.000000000","updated":"2020-10-17 21:45:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e92beb4e3618bd2e4303ffb7b0f4516e67a82743","_number":758651,"virtual_id_number":758651,"owner":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-17 12:44:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":-2,"date":"2020-10-17 21:45:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-17 12:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-17 21:45:01.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"7e799a0c761bb402845aba10787fcdb62b301d98","author":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"date":"2020-10-17 10:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e45d2158515a87eea501c560a414df83a841e8c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-17 12:44:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d30f162b727b4d5e89d6d316af4691c7 : FAILURE in 7m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e773e58a8b9e454ea1c175cb84c376e3 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e57308582af4fb983c9320584672e8b : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1847d997588b49309b224ac8f430a330 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9860fcb6a3584be1a389540823cfcf30 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55eacb6a40e4c5d945e6478a8956735 : SUCCESS in 6m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81c69d89da864505b158ae24920bd354 : SUCCESS in 6m 29s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/96a6d3824e9e4e04ad89b224f463f5e2 : TIMED_OUT in 2h 14m 23s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fc2bac7c9f54cdaa43de49b8ecc2d66 : FAILURE in 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8649e8fa10fcff58aced89377b2120201d7daec2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-17 21:45:01.000000000","message":"Patch Set 1: Code-Review-2\n\nDo not make such cosmetic contribution to open source project.","accounts_in_message":[],"_revision_number":1},{"id":"010e50e73270b5259505a5daf23d081416e8820f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-17 21:45:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbac8f7dd673e7da5dcce62fadbfc8d5836d132c","revisions":{"cbac8f7dd673e7da5dcce62fadbfc8d5836d132c":{"kind":"REWORK","_number":1,"created":"2020-10-17 10:24:28.000000000","uploader":{"_account_id":30696,"name":"zhaoleilc","email":"15247232416@163.com","username":"zhaoleilc"},"ref":"refs/changes/51/758651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/758651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/758651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/758651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/758651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/758651/1"}}},"commit":{"parents":[{"commit":"349c63233558e70f8cdc97a347c3bae6a9e3cd6e","subject":"Merge \"Fix error handling on instance create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/349c63233558e70f8cdc97a347c3bae6a9e3cd6e"}]}],"author":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-10-17 10:19:40.000000000","tz":480},"committer":{"name":"zhaoleilc","email":"15247232416@163.com","date":"2020-10-17 10:24:22.000000000","tz":480},"subject":"Fix a typo in the document","message":"Fix a typo in the document\n\nThis patch changes \u0027you dont want\u0027 to\n\u0027you don\u0027t want\u0027 in integration/README.md\n\nChange-Id: I4cfddd6085f0ce4605585563a3e043b2d85ad5b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cbac8f7dd673e7da5dcce62fadbfc8d5836d132c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/cbac8f7dd673e7da5dcce62fadbfc8d5836d132c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
