)]}'
{"id":"openstack%2Ftrove~758819","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~I8fb00ae6c0c2a400f78b97b9d0bfbf53ac297d5b","project":"openstack/trove","branch":"stable/victoria","topic":"victoria","hashtags":[],"change_id":"I8fb00ae6c0c2a400f78b97b9d0bfbf53ac297d5b","subject":"Use victoria branch for CI and image build","status":"MERGED","created":"2020-10-19 13:54:29.000000000","updated":"2020-10-19 22:59:42.000000000","submitted":"2020-10-19 22:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758819-1603148143304-d11e6423","meta_rev_id":"7af913001f3751097884d118fcda5d3d738682da","_number":758819,"virtual_id_number":758819,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 22:55:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-19 20:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-19 20:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 22:55:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf6172c57469c84a5f9ace450e5963ccfcbadb1d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-19 13:54:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b8583900a8215406f8f1946c6552ff99fa71cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 16:19:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d6b6cacf25d49b1a7ad800798986f5c : FAILURE in 10m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/575eb581a6e64e9ca413d961ebb548a3 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e6af59418249b6910358c9c9905eab : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed5fe5d5d09747d28d7bc8ec97e424b5 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ce1ce0d1fff45bfbbc055ac00befe9b : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ce0604adde4fb49844d4c1bf41851a : SUCCESS in 7m 47s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29f815d2d35249f58dc4736307be73f4 : SUCCESS in 7m 12s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1305c98078c84a1bb3ff62e00ce80ff0 : TIMED_OUT in 2h 14m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1bf8875e3a84d6aa3967e17156e1459 : SUCCESS in 1h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db642d7c81f9d91217dcb039c77d4d9600dfa721","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-19 20:58:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d571fd812372811ea7a8a591ed07ceec74814b17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:59:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f1ff1d9cc4b72a752a62bb7b0d2f8b19218fd86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:55:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57ae5b895bbc4d1e89fc5c863459f351 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba1f6f7bc054ef19d67a7b43a5b223f : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44be01ffc79347f69ab140dfa92bb78c : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bcd821c588e48ddb13a911a24bd8fe2 : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b2aa084fea4243815119de9050696f : SUCCESS in 10m 22s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/577e443ab25348719ceeaf15e51fb47f : FAILURE in 1h 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60a4ac66c1a0d00d5fd55200ac1302cf25343324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:55:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"01660e6181890b95d64e3abf6909ed7828f26c40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 22:59:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f61bdb26074904b17a1c682c68b6ad : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfb534c1f2f2f9bcec30a2b056195e0ba067a63d","revisions":{"dfb534c1f2f2f9bcec30a2b056195e0ba067a63d":{"kind":"REWORK","_number":1,"created":"2020-10-19 13:54:29.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/19/758819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/758819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/758819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/758819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/758819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/758819/1"}}},"commit":{"parents":[{"commit":"555ddb030e4f3e98e37f92fbb4892a15bf919261","subject":"Create floating IP in the user\u0027s project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/555ddb030e4f3e98e37f92fbb4892a15bf919261"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-19 13:53:55.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-19 13:53:55.000000000","tz":780},"subject":"Use victoria branch for CI and image build","message":"Use victoria branch for CI and image build\n\nChange-Id: I8fb00ae6c0c2a400f78b97b9d0bfbf53ac297d5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dfb534c1f2f2f9bcec30a2b056195e0ba067a63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dfb534c1f2f2f9bcec30a2b056195e0ba067a63d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
