)]}'
{"id":"openstack%2Ftrove~759127","triplet_id":"openstack%2Ftrove~master~Icd6345d6c97648cdfbfaa8d9edac7315a1409356","project":"openstack/trove","branch":"master","topic":"list-instance-by-project","hashtags":[],"change_id":"Icd6345d6c97648cdfbfaa8d9edac7315a1409356","subject":"Support to get instances of a specified project","status":"MERGED","created":"2020-10-21 19:37:28.000000000","updated":"2020-10-22 11:25:44.000000000","submitted":"2020-10-22 11:24:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759127-1603365850844-e5953300","meta_rev_id":"287a75e7be6361d558a76522c443d06b2eb9786d","_number":759127,"virtual_id_number":759127,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 11:24:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-22 09:09:55.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-22 09:09:55.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 11:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f065c66b65bf46c75891834795aef03edf04fb25","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-21 19:37:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efb86205456efeed8e0b77a57d707cafdba230f4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-21 20:53:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70bf9c3b708c6ffe03d682e042ea25d87fa75595","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-21 20:54:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e378d4534c43cced7ed2248541bd4611233b1492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 23:25:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27b5fc9e0a7848bcb604ffef646d2677 : SUCCESS in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/686b6fcc9d1b4274b75449b0cc589b99 : FAILURE in 7m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8203b34f79e741199fcffe0b25f7ee3f : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb401126c7a446dbe3f3c406e05410c : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a8ca2ce8c1a4f848197e0bdef3f34b8 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/100119eceb5f43b28b2f15e070a57997 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23464e8f642145fca47f89a98ea7f032 : SUCCESS in 7m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c2b3c05f64643e9ac172b2414fafaea : SUCCESS in 8m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/be078b9cf58e44288ad782e5686d07f4 : FAILURE in 1h 59m 09s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d43472b9b55e4f45ae73210ba47ee1fa : SUCCESS in 1h 09m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee757200e49dca11b5ba1a9a810becf37c67641a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-21 23:31:14.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4c86e5d8b04415cb5f41e190d37045ae36ec8ce2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-21 23:32:10.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 0ce59e0436230cc19083afca63928d0fdfad7fcc","accounts_in_message":[],"_revision_number":4},{"id":"4a5d994b5b99fa24d740adb1fb0d8cf3cb0bdc06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 02:01:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ddbe22a62084f9da8dcae2b110ffd69 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb21a8021a7344068dbf5c0d5b6d8398 : FAILURE in 9m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437729bf7b394c7f91b66412b2656759 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec80df3624194a2ab114542e60bc15d2 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aef5800fab64730b2fff6ee4ff288b7 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf6ff0fd6864f4ab50fc9f5e73844c9 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c4d8c30e4146888c2d8b88b19c2b7d : SUCCESS in 7m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d4f600abc504fe99d98732f962ac132 : SUCCESS in 7m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a9ac358dabc94cce9165bbc9cf0481cd : FAILURE in 2h 01m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/556e20a19ae446078222396a94f2d432 : RETRY_LIMIT in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f45539cfd70c42f61ec6793163e59f94c0319713","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 09:09:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"25c98aad2a642b575c89b71439a756223838e56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 09:10:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1ad8013e2e0243569e3346a9dcb4c2927906c2b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:24:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"10219a5694e41755ef91ee1b9c8948d3174062a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:24:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b67bf5d3ce834d15b5bab8c7ba620115 : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7f8d924773a4f7e89fc070d39d987df : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7b06536d3f49bbb01d32df13343af2 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/623b69eb52b74e738b77a457fb7f933e : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f646d174b2ef494ab9fb0117e5d14397 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b89511634ec4450829fbf71e016fdae : SUCCESS in 8m 52s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/03cb218bf5d6473c916de2abb3acac8f : FAILURE in 1h 48m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bceeaec35bffcdf75c50e11406c8326469ff941d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:25:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ec841f77d11b4a9197e16fe80c54bf77 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/001b9b7ef0334e5eb8f05c82fb6ad1d0 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bfa2392d16a803e1d95edfff125b2ef0b4db1ec2","revisions":{"e5514baeb49ada831c6d5314bf3daf208f70e097":{"kind":"REWORK","_number":1,"created":"2020-10-21 19:37:28.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/27/759127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/759127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/759127/1"}}},"commit":{"parents":[{"commit":"6a12e596012232d5d25c027cf4ef5335c00def83","subject":"[TrivialFix] Fix \u0027--replica-of\u0027 argument format in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a12e596012232d5d25c027cf4ef5335c00def83"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 09:25:12.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 19:36:52.000000000","tz":780},"subject":"Support to get instances of a specified project","message":"Support to get instances of a specified project\n\nChange-Id: Icd6345d6c97648cdfbfaa8d9edac7315a1409356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5514baeb49ada831c6d5314bf3daf208f70e097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e5514baeb49ada831c6d5314bf3daf208f70e097"}]},"branch":"refs/heads/master"},"13f1845c4de25c577b83b10d8ac0114e6f7fa7e8":{"kind":"REWORK","_number":2,"created":"2020-10-21 20:53:06.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/27/759127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/759127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/759127/2"}}},"commit":{"parents":[{"commit":"6a12e596012232d5d25c027cf4ef5335c00def83","subject":"[TrivialFix] Fix \u0027--replica-of\u0027 argument format in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a12e596012232d5d25c027cf4ef5335c00def83"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 09:25:12.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 20:44:32.000000000","tz":780},"subject":"Support to get instances of a specified project","message":"Support to get instances of a specified project\n\nChange-Id: Icd6345d6c97648cdfbfaa8d9edac7315a1409356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13f1845c4de25c577b83b10d8ac0114e6f7fa7e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/13f1845c4de25c577b83b10d8ac0114e6f7fa7e8"}]},"branch":"refs/heads/master"},"ba67fd0aabe00e5d859d8b96a4a40fda474e3ac0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-21 20:54:39.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/27/759127/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/759127/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/759127/3"}}},"commit":{"parents":[{"commit":"6a12e596012232d5d25c027cf4ef5335c00def83","subject":"[TrivialFix] Fix \u0027--replica-of\u0027 argument format in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a12e596012232d5d25c027cf4ef5335c00def83"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 09:25:12.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 20:53:08.000000000","tz":780},"subject":"Support to get instances of a specified project","message":"Support to get instances of a specified project\n\nThis patch also fixes a performance issue when the adm user is getting\nall the instances.\n\nChange-Id: Icd6345d6c97648cdfbfaa8d9edac7315a1409356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba67fd0aabe00e5d859d8b96a4a40fda474e3ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ba67fd0aabe00e5d859d8b96a4a40fda474e3ac0"}]},"branch":"refs/heads/master"},"bfa2392d16a803e1d95edfff125b2ef0b4db1ec2":{"kind":"REWORK","_number":4,"created":"2020-10-21 23:31:14.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/27/759127/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/759127/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/759127/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/759127/4"}}},"commit":{"parents":[{"commit":"6a12e596012232d5d25c027cf4ef5335c00def83","subject":"[TrivialFix] Fix \u0027--replica-of\u0027 argument format in docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6a12e596012232d5d25c027cf4ef5335c00def83"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 09:25:12.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-21 23:31:01.000000000","tz":0},"subject":"Support to get instances of a specified project","message":"Support to get instances of a specified project\n\nThis patch also fixes a performance issue when the adm user is getting\nall the instances.\n\nChange-Id: Icd6345d6c97648cdfbfaa8d9edac7315a1409356\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bfa2392d16a803e1d95edfff125b2ef0b4db1ec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bfa2392d16a803e1d95edfff125b2ef0b4db1ec2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
