)]}'
{"id":"openstack%2Ftrove~759240","triplet_id":"openstack%2Ftrove~master~Id85c25c0cd3903f1b536a08ddcf69e306ce7345d","project":"openstack/trove","branch":"master","topic":"user-port","hashtags":[],"change_id":"Id85c25c0cd3903f1b536a08ddcf69e306ce7345d","subject":"Fix getting user port for instance","status":"MERGED","created":"2020-10-22 10:16:31.000000000","updated":"2020-10-22 21:33:05.000000000","submitted":"2020-10-22 21:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759240-1603402265336-902aecb7","meta_rev_id":"28a68ec80d28dd6ce3e4e7991064314d7d5bafdf","_number":759240,"virtual_id_number":759240,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 21:31:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-22 19:34:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-22 19:34:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 21:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"609df2452426f0e56933b459b9f61e579658a447","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 10:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e052997fbbb22a665572bed0b33f220a21cfa15","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 10:17:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f42b9b8f5fcf22227108208ff9ceb3c3c8977de6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:05:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/185200920be74733bae88c564c6e5368 : FAILURE in 10m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19eaed1ca48c402587605c5125ad4d9b : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5041216f38a0468daf6aeefa60be1894 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7f9578502d842eba6acb03eadeb7414 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dcf5b9441f04e46a2549d5bf78ce741 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75a2eb28fed34d41b76e44d0cc66ff7e : SUCCESS in 7m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21b687cf1ec444f0bec58854060fead3 : SUCCESS in 8m 15s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d469b34e4e4b4656aa6ec316fdb6ee62 : FAILURE in 1h 45m 45s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b753bb9e6841f3bbe83644b40cc81b : FAILURE in 1h 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b1d5cfedde50941d919db4bba5b1dff4afe1b05","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 19:34:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d8060f03c159635bb139e9ecb3b5f5351f6cbb8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 19:34:39.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 666b83aeab13af3dd54895a25ffd8816a51c888e","accounts_in_message":[],"_revision_number":2},{"id":"92571119eb0cbb37995a5c202b756858c95ae3ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 19:34:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebdcfa032b931158d429467ffd3499ac1b8c62c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"23202220a70a0a101a1bdc4855a5a529b310d07e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:31:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb2b85b16c1e4ac5b3d15928158005c0 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b76c87efaab94d6ca921cd1c9412e852 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80632de72dff49dbb405d79d08d4e7c7 : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c41caf2c02a4dff9112863bd8e7cd2a : SUCCESS in 15m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5db84293aa4bd99af9504f2df28186 : SUCCESS in 8m 43s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/327dc76c7e6843bc821dedef08ccd06b : FAILURE in 1h 49m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a55c8b9c19a34d1fae5b103fc0af1e8ee952898","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:33:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab0038bb97d4e838be6b3ab5031f34b : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d734037cb9a53a55df9bc819b48234bf412d13c9","revisions":{"77e5deb36ba75070dbd6ca972b5f549ea58c8bde":{"kind":"REWORK","_number":1,"created":"2020-10-22 10:16:31.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/40/759240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/759240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/759240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/759240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/759240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/759240/1"}}},"commit":{"parents":[{"commit":"060a100aa069ef8e1e15553dea868edd8aa614cb","subject":"Merge \"Remove the unused coding style modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/060a100aa069ef8e1e15553dea868edd8aa614cb"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-22 10:12:50.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-22 10:12:50.000000000","tz":780},"subject":"Fix getting user port for instance","message":"Fix getting user port for instance\n\nChange-Id: Id85c25c0cd3903f1b536a08ddcf69e306ce7345d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77e5deb36ba75070dbd6ca972b5f549ea58c8bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77e5deb36ba75070dbd6ca972b5f549ea58c8bde"}]},"branch":"refs/heads/master"},"d734037cb9a53a55df9bc819b48234bf412d13c9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-22 10:17:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/40/759240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/40/759240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/759240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/759240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/40/759240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/40/759240/2"}}},"commit":{"parents":[{"commit":"060a100aa069ef8e1e15553dea868edd8aa614cb","subject":"Merge \"Remove the unused coding style modules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/060a100aa069ef8e1e15553dea868edd8aa614cb"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-22 10:12:50.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-22 10:16:34.000000000","tz":780},"subject":"Fix getting user port for instance","message":"Fix getting user port for instance\n\nCheck the port type by network rather than port description.\n\nChange-Id: Id85c25c0cd3903f1b536a08ddcf69e306ce7345d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d734037cb9a53a55df9bc819b48234bf412d13c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d734037cb9a53a55df9bc819b48234bf412d13c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
